)]}'
{"id":"openstack%2Freleases~726207","triplet_id":"openstack%2Freleases~master~I2abfe56317727369963c6c7fa7580e89c5b1e07c","project":"openstack/releases","branch":"master","topic":"ovn-octavia-provider-0.1.1","hashtags":[],"change_id":"I2abfe56317727369963c6c7fa7580e89c5b1e07c","subject":"Release ovn-octavia-provider 0.1.1 for stable/ussuri","status":"MERGED","created":"2020-05-07 17:02:53.000000000","updated":"2020-06-29 09:24:13.000000000","submitted":"2020-06-29 09:24:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726207-1593422653236-0147cf3e","meta_rev_id":"41a284cecd593ceae04c42f5aa44a68413ef53d2","_number":726207,"virtual_id_number":726207,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 09:24:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-29 09:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-06-26 07:55:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-06-26 05:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-06-26 08:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-06-26 14:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-29 09:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-06-26 07:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 21:23:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-26 05:58:58.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-06-26 07:55:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-26 08:03:13.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-06-26 14:26:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-29 09:09:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-06-29 09:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2efecf0feac455bfcc6c30b5c30e68c45edcf78","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-07 17:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8353d178db5281bf08d4569bb8bc18f2e322b4c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-07 17:04:18.000000000","message":"Patch Set 1: Workflow-1\n\nSetting -W until I can ping Maciej, just wanted to get this out.","accounts_in_message":[],"_revision_number":1},{"id":"ade2d4807cb46cf42256be207419edbad15caf2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:15:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5d98d4966a4071b0e9aa58ff4e46c8 : SUCCESS in 10m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/394373f864ef49f5bb6978d492baf0bc : SUCCESS in 6m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5dc795e03e1d4dae8e9165c170b9cccf : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5cf49acbbce4deb5e14800dd0205aab5383bff45","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 12:20:18.000000000","message":"Patch Set 1:\n\nStill working on things for this Brian?","accounts_in_message":[],"_revision_number":1},{"id":"0ae8729f8a26dad0ebb6c8c154f138a229cee869","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-22 12:56:45.000000000","message":"Patch Set 1:\n\nI think we can update this to point: 2448c4925b7c2b2e4f4287a14bc67fa5e8360002 https://review.opendev.org/#/c/734770/ and release now.","accounts_in_message":[],"_revision_number":1},{"id":"a3034550ea67a65f75e6e92f04f2a6b35369e499","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-23 23:30:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"41aac4dfc534c60c88e40f36296acdeccf27d74f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-23 23:31:22.000000000","message":"Patch Set 2: Code-Review+1\n\nUpdate the release hash to the latest commit in stable/ussuri.\nLooks good to change the type to \"other\" too.","accounts_in_message":[],"_revision_number":2},{"id":"f12ef37443860710366b473a86eecbe53a5197b8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 23:31:52.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a78fe83c4e8ef0043e627b52cf1ffb1d91b47b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 23:38:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30bce965ee524ce881c4da4dbefb0961 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdd4f06760d743389202e0e5af001ec9 : FAILURE in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/516548634d794cd89c4ac92a443e206f : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6e76fcb00de3da2d658b16d34e9a35cb52c830d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 21:23:00.000000000","message":"Patch Set 2: Code-Review-1\n\n0.2.0 release was already done in victoria. This will need to be a 0.1.1 release.","accounts_in_message":[],"_revision_number":2},{"id":"934ea70f3e2c17e28d774f27550de5190fadc0e3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-24 22:40:02.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"c4a3c5b26affd969ac74ee50a2bc5247b08ffdea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-25 18:06:05.000000000","message":"Patch Set 2:\n\nSean - I\u0027ll send an update to make this 0.1.1, but in reality it should be 0.2.0 and Victoria should have been 1.0.0","accounts_in_message":[],"_revision_number":2},{"id":"2a5024ceb66b17c281a13871d35159f1c25ffa5c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-25 18:08:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a670352bd9a886f3a88b9cbe57e47d003e63cac6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-25 18:08:58.000000000","message":"Topic changed from ovn-octavia-provider-0.2.0 to ovn-octavia-provider-0.1.1","accounts_in_message":[],"_revision_number":3},{"id":"f5c3d794304054d5dccfba3493b3897663071b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:26:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92d553c0b4745e3892d16de6d203e03 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f219354773d4f689e9c32a5a039dd53 : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b1b0ac7e8ccf41af914c3c731ea94635 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"123e8d2ef5af5d7a6909df6dce493afc3b9caf1d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-26 05:58:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d660305c1cee0d4d1751be0f8e3ff114144e97bc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-26 07:55:49.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2a4204a98a9b656250b8c7d9fefd67b2d50d9312","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 07:57:14.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"afaa6574f2b7036604b2fd7e528ed9c3dd7bd37d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-26 08:03:13.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e but in reality it should be 0.2.0 and Victoria should have been 1.0.0\n\nYeah, I totally forgot it when looking at 0.2.0 release in Victoria :(","accounts_in_message":[],"_revision_number":3},{"id":"1ba25e092ddf97317ada9eaf4752f5e9e880b650","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-26 14:26:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"297da2bcb38bde53e7cba308cb7c992df514d63b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-29 09:09:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"01e2ed7da05fe7ee899081df3cc9407f18bb9fc5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-29 09:09:36.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"98a1c52be6cd688c4f82a7b717b42089f0660d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:09:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c9bab28bbc937241670ab1b56119541da9775a7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:24:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8bda3d0cd24446284498d2a5ae52e52 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f57a8b8fc048cd96835544e212f125 : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a35c1d0203a4429dbcf7d5a639932eb1 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"578467cf0d1087aaf12c30f93c6f3cfbb26cf99f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:24:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8037304c2d876aa2c68a1ae996b5b291b1ba078a","revisions":{"96250a5cd80e509189fb7d51d89ec8b621652e65":{"kind":"REWORK","_number":1,"created":"2020-05-07 17:02:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/07/726207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/726207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/726207/1"}}},"commit":{"parents":[{"commit":"44595097c653ad4549cda9651b81944e3a513589","subject":"Merge \"Release nova RC2 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44595097c653ad4549cda9651b81944e3a513589"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-05 15:06:38.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-05 15:06:38.000000000","tz":-240},"subject":"Release ovn-octavia-provider 0.2.0 for stable/ussuri","message":"Release ovn-octavia-provider 0.2.0 for stable/ussuri\n\nI also changed the type to \"other\" as this is not a library.\n\nChange-Id: I2abfe56317727369963c6c7fa7580e89c5b1e07c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96250a5cd80e509189fb7d51d89ec8b621652e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96250a5cd80e509189fb7d51d89ec8b621652e65"}]},"branch":"refs/heads/master"},"f8113462aec8b57e09f0cbf1d239387853899368":{"kind":"REWORK","_number":2,"created":"2020-06-23 23:30:21.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/07/726207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/726207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/726207/2"}}},"commit":{"parents":[{"commit":"44595097c653ad4549cda9651b81944e3a513589","subject":"Merge \"Release nova RC2 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44595097c653ad4549cda9651b81944e3a513589"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-05 15:06:38.000000000","tz":-240},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-06-23 23:29:20.000000000","tz":0},"subject":"Release ovn-octavia-provider 0.2.0 for stable/ussuri","message":"Release ovn-octavia-provider 0.2.0 for stable/ussuri\n\nI also changed the type to \"other\" as this is not a library.\n\nChange-Id: I2abfe56317727369963c6c7fa7580e89c5b1e07c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8113462aec8b57e09f0cbf1d239387853899368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8113462aec8b57e09f0cbf1d239387853899368"}]},"branch":"refs/heads/master"},"8037304c2d876aa2c68a1ae996b5b291b1ba078a":{"kind":"REWORK","_number":3,"created":"2020-06-25 18:08:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/07/726207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/726207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/726207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/726207/3"}}},"commit":{"parents":[{"commit":"44595097c653ad4549cda9651b81944e3a513589","subject":"Merge \"Release nova RC2 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44595097c653ad4549cda9651b81944e3a513589"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-05-05 15:06:38.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-25 18:08:19.000000000","tz":-240},"subject":"Release ovn-octavia-provider 0.1.1 for stable/ussuri","message":"Release ovn-octavia-provider 0.1.1 for stable/ussuri\n\nI also changed the type to \"other\" as this is not a library.\n\nChange-Id: I2abfe56317727369963c6c7fa7580e89c5b1e07c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8037304c2d876aa2c68a1ae996b5b291b1ba078a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8037304c2d876aa2c68a1ae996b5b291b1ba078a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
