)]}'
{"id":"openstack%2Freleases~726701","triplet_id":"openstack%2Freleases~master~I15be9a12ff6684729980c40afae9625a8b8e550a","project":"openstack/releases","branch":"master","topic":"validate-timeout","hashtags":[],"change_id":"I15be9a12ff6684729980c40afae9625a8b8e550a","subject":"Speed up sdist build","status":"MERGED","created":"2020-05-10 22:45:54.000000000","updated":"2020-05-11 10:55:14.000000000","submitted":"2020-05-11 10:55:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726701-1589194514968-b84da92b","meta_rev_id":"6f7927ceddf0f8c6c7c0de9f2502ded2c165f42a","_number":726701,"virtual_id_number":726701,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 10:55:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-05-11 09:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-05-11 10:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-05-11 10:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 09:27:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-11 10:41:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-11 10:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f59ba8c8e97aa40cc2eaa32fd6380c9d3c4ed42b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-10 22:45:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe3110f2063f490bb490e4b4c41ab86719728a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:58:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7da7f1388848dcb8326a2723618e54 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/847299d1e0c840e6ab9e910ec488b637 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01dd55e662b348ceb07c1ff3b57e90b7 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52a1ff6553e348a9b03de9e97b3a300b : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d1b1cb21e140e7a2312a82f4bee0ca : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b8a731323894ed9b324a1b217936dda : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0fc0d2ef97a44b12b9607c0b3f930583 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"80dbbf45ca3e78f7db176d01ec231c98cc1e274e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-11 09:27:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d4d4037dc030fb1dce10ff3c681cffc302b3767","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-11 10:41:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da9c4831215477662ed5d596f7584ea0fcb9178d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:41:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99f7f611c28066b681e40100082d6bf751577432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:55:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da3e1ae193e24d5934f21bf296afe1d963c2867b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:55:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15dae1b8297546d4b3aecb4c534e95b3 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b7e6b3538a4a11b2de2518ae30e5e6 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7458e5e1d044470a99180b61b3912c5c : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5616bd730047ebb4f77ea9c4faafc8 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6547824bac2944d4b3fc4ac61f3808ca : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"854d9e270bf45f45b48d960a983b8f8b9e94cb0a","revisions":{"854d9e270bf45f45b48d960a983b8f8b9e94cb0a":{"kind":"REWORK","_number":1,"created":"2020-05-10 22:45:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/01/726701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/726701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/726701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/726701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/726701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/726701/1"}}},"commit":{"parents":[{"commit":"b19a92adbedb4b5d071abce9bbec589ef50603df","subject":"Install wheel in validate venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b19a92adbedb4b5d071abce9bbec589ef50603df"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-10 22:42:35.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-10 22:42:35.000000000","tz":-300},"subject":"Speed up sdist build","message":"Speed up sdist build\n\nWe have a couple validations that need to build an sdist now. But once\nit has been built in our temporary pristine clone, we don\u0027t need to\nbuild it again. Because we create a clean clone, we also don\u0027t need to\nworry about any stale dist builds. This adds a check to the sdist build\nlogic to recognize if an sdist has already been built and skip it.\n\nThere are also a few extra steps in the build process that are not\nneeded for our purposes. The biggest is the new(ish) pbr behavior of\nbuilding all release notes in the git tree if reno is present. This adds\nan environment variable flag to tell pbr not to do that. While we\u0027re at\nit, we don\u0027t need AUTHORS or CHANGELOG created either, so set the flags\ntelling pbr to skip those as well.\n\nChange-Id: I15be9a12ff6684729980c40afae9625a8b8e550a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/854d9e270bf45f45b48d960a983b8f8b9e94cb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/854d9e270bf45f45b48d960a983b8f8b9e94cb0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
