)]}'
{"id":"openstack%2Freleases~726838","triplet_id":"openstack%2Freleases~master~Ia47e73b51b3216572d51399338e95bd4269a8d1d","project":"openstack/releases","branch":"master","topic":"sdist","hashtags":[],"change_id":"Ia47e73b51b3216572d51399338e95bd4269a8d1d","subject":"Pass full environ copy to sdist subprocess","status":"MERGED","created":"2020-05-11 12:56:21.000000000","updated":"2020-05-12 09:21:38.000000000","submitted":"2020-05-12 09:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726838-1589275298918-5004bf0a","meta_rev_id":"4ccc764c9b533b9b8fe48b4a91282a587bac8cec","_number":726838,"virtual_id_number":726838,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 09:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-12 09:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-12 09:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 09:10:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-12 09:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d0cea86e2b538bb64bde1e81988cd9914bac094","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 12:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7220dd728d1e00446d5ed87dd24daf84e54f2893","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:58:58.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"351ca3fda3c85879ae8cb247880c36aedf52e896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:23:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24001ebf1e34156b353ff06237ea9c6 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59473e4cc46649dbbd3f6725742d7431 : RETRY_LIMIT in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a394da5ed59d444f871d5132c85477a4 : RETRY_LIMIT in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80219b0000224ce284246634efa05399 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba5896798ef4fdea3339e918da3b246 : SUCCESS in 9m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3add2b009b574bd0854f6aca0f7e3bdd : FAILURE in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19ae8e4118744f1da73b88455ae8483a : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"39f78de8946822f268787b0fbd9af75de40f1a68","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 13:24:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb82cbb0ba053290d819d37547bd0618eea1acb2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 13:29:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32b6376542099e60d5ef2dcbd3b90071131b9e7b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:38:44.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"22673b5c0d1da053657e77476413fbeca3af600b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 14:42:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6e102d12cb8132307d7ee8ecfbaf894133b4b7bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 14:47:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f7063aa8556432347386fcea76648abd4155667","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:51:11.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a64cd98178e3e824ec34fd3015674b30eb52e8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:01:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ea5ba29a0247e294260a765075e1bc : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3bd489e153f4650a3ba3271173a91c2 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38b4365d9c934e95ad7ab01ef924498f : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3d0d4de245843ef952223accf4e685b : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8deb8ce44451428296c804433594700e : SUCCESS in 9m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ec0a562f50f496f9be06b06f1bc0bcc : FAILURE in 7m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/762e2bf2cabf4d10b5451c0709a3b6c4 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"00033801ec5948041d8049afe9f52296c88b88b4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-11 15:37:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"1401b5390565cc3f052aa951ddf80c8e4c972425","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 15:48:13.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"26037337351bc8e39e2bd9fc98d1e34e8a8f28b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 15:48:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc8e643fff9bcb8872a183f7fe4b820b0c74724b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:49:41.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"5dba413e0b72a61efd002e1a517ab9517255ffed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:01:58.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189f1d9d9a2d4f588376acf43402e65b : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b4268971464f05a34f4b49c2eb340c : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d30caaac2154ad9a466cff7d0a49bdf : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0706b3be660407a92c91b9906234a33 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfbbdd32f124e52ba503488ea691936 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a5c254132194db6bc33b8571882e9de : FAILURE in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/95fd2fdf43ae4571b450318d2b11c926 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":4},{"id":"1e75e17fc5298a0cecf4d4ab46512bf053c8e07f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 16:06:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f38424e40880710de93549cbb52855d59585662a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:08:39.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"6d186abd1aa96ef66583b3dd9921f191a61cd22b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 16:16:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"13779c2d7c8d17906cdeddf5a611a721dd9d806c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:20:10.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"83fbe6984caba465611fe2b658a71def86b6931f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 16:31:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b27a51a6903432e814e60d67c3760263578eae93","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:34:35.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"746fcd7e7cb55f6efa743f8e673e84d91c420740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:45:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de91925c3b8f4adabbc162ebb3692e8f : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd3ed6ad82e643169c2596c020e3d70e : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ef9183bacca4791941fb79e37050530 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d000f0f8a7db40e495a1123c4cbac37b : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c287858eba4137a09719617bd2dfe3 : SUCCESS in 11m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a59b11340414449ac742d5ed3d67435 : FAILURE in 6m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a0798747c97425b90d656e5b98d3bc7 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":7},{"id":"c7620d756c26ee8eab977153044dfff6e2eb01a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 16:52:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"765f566ef6de958cd74b9f5627544ab066f3e37e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:55:38.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"4b52dda4208ced3eedfacf10d2b50bbf738d159c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:08:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b755b95b344b80b5b016f0d4f931c6 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa6b7bcd5f8e4110bd5ed00e15004f20 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/919908fbe3f742068effe3dd1d4ef353 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b4e05fdc9b840409e2f59f9084868c1 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104d10a56d4d4be0b6721ef7cf646865 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdd51723add247ae9fa238222290e1b1 : SUCCESS in 6m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9d199da543f4cc1bb13faa07182dac2 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":8},{"id":"1ef483db636b622aaa7090d714d6b304c5cec1d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 18:01:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4e8f8c07b592acf2baed01c648c1627eee0ebd25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:14:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18509db54e24ee68832ae09bea57786 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa8a97de9f84f0195f1027ca054e552 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f2260a59e1041b79a2a15b0874a44e3 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4de05740484840b11b535a86558655 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e34a026f80348d3b50fd46b907810fa : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ce833d5a50947eb81aafc0e0c19b08f : SUCCESS in 5m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/09e2d81bbca44380bb260135ae60d6dd : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":9},{"id":"a947488d3d0eb038c22f90c18187f23c93104432","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 22:09:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6d65df87f1ef3fe87131780d6d11a1897a5b1fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:19:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca88fa27f214b26946d5576f36d8615 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c7d4b90eca543b986b34005f0ee6f71 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3a5d29322f7426f935a31fc305d9930 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98292ed643334c5f9dd1aec7f5ce25d5 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f9670fb1974c3fb000a3bd440eac28 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13bab04878d5439eba3ed98f1d96e740 : SUCCESS in 6m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/723d0e97a14649dfbeb116b270d41c3b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":10},{"id":"83e83e9c5876d23b16d919220acedc1f95d7933a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-12 09:10:09.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f0f81e420c34cde1f0f05b810f87015e2d134993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:10:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"3fe25fc7111667f0c8dcf2b145c7ad3d27a2e455","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:21:38.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af49159a7ff946e68e3267ce8af7edf4 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7607d909ad53487089de99320b1f25fe : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10c5fde06f514421a07a9ac354aaccf9 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a66396dd914aa8b330b62daf7fa99c : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98c97daee7f44f04a35e96a08a67d03b : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":10},{"id":"d4aa79931c9a747f210cdcf5ece8c9246dfa6934","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:21:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f402cde902143bfb06641f3154ad834c08218cef","revisions":{"c713f76b18c56eecfa20d64b23b3d7370795a197":{"kind":"REWORK","_number":1,"created":"2020-05-11 12:56:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/1"}}},"commit":{"parents":[{"commit":"b36a598529d9e3ce8c69978b7f5b941f735368c5","subject":"Merge \"Speed up sdist build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b36a598529d9e3ce8c69978b7f5b941f735368c5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:56:15.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c713f76b18c56eecfa20d64b23b3d7370795a197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c713f76b18c56eecfa20d64b23b3d7370795a197"}]},"branch":"refs/heads/master"},"67736537040254349b9c09010160a76c2488c966":{"kind":"REWORK","_number":2,"created":"2020-05-11 13:29:25.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/2"}}},"commit":{"parents":[{"commit":"b36a598529d9e3ce8c69978b7f5b941f735368c5","subject":"Merge \"Speed up sdist build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b36a598529d9e3ce8c69978b7f5b941f735368c5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 13:29:19.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67736537040254349b9c09010160a76c2488c966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67736537040254349b9c09010160a76c2488c966"}]},"branch":"refs/heads/master"},"289ed4eef9393285307e1f5af3345ec7938ba7c2":{"kind":"REWORK","_number":3,"created":"2020-05-11 14:42:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/3"}}},"commit":{"parents":[{"commit":"b36a598529d9e3ce8c69978b7f5b941f735368c5","subject":"Merge \"Speed up sdist build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b36a598529d9e3ce8c69978b7f5b941f735368c5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 14:42:00.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/289ed4eef9393285307e1f5af3345ec7938ba7c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/289ed4eef9393285307e1f5af3345ec7938ba7c2"}]},"branch":"refs/heads/master"},"f7341ea2d20949a583356011fd15b177226f7fa3":{"kind":"REWORK","_number":4,"created":"2020-05-11 15:48:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/4"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 15:41:41.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7341ea2d20949a583356011fd15b177226f7fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7341ea2d20949a583356011fd15b177226f7fa3"}]},"branch":"refs/heads/master"},"af652c8d6fb52942428b9861281ace8855bf94ea":{"kind":"REWORK","_number":5,"created":"2020-05-11 16:06:24.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/5"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 16:06:19.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af652c8d6fb52942428b9861281ace8855bf94ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af652c8d6fb52942428b9861281ace8855bf94ea"}]},"branch":"refs/heads/master"},"675e08e1c057ec4468071d77fba5a2eaf9692869":{"kind":"REWORK","_number":6,"created":"2020-05-11 16:16:40.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/6"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 16:16:35.000000000","tz":-300},"subject":"Revert sdist rerun check","message":"Revert sdist rerun check\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/675e08e1c057ec4468071d77fba5a2eaf9692869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/675e08e1c057ec4468071d77fba5a2eaf9692869"}]},"branch":"refs/heads/master"},"29f8d6fcaa9f7b0baa2c9b6fbdeaec8aa435fa80":{"kind":"REWORK","_number":7,"created":"2020-05-11 16:31:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/7"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 16:31:11.000000000","tz":-300},"subject":"Pass full environ copy to sdist subprocess","message":"Pass full environ copy to sdist subprocess\n\nIn order to tell pbr not to build release notes and create CHANGELOG and\nAUTHORS files, we need to pass flags in as environment variables. We\ncan\u0027t just pass these flags though, as that will exclude the rest of the\nnormal environment set in. So copy the current environment, append our\nupdates, and pass that in explicitly to the subprocess call.\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/29f8d6fcaa9f7b0baa2c9b6fbdeaec8aa435fa80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/29f8d6fcaa9f7b0baa2c9b6fbdeaec8aa435fa80"}]},"branch":"refs/heads/master"},"04563c8ee317a89a26a46d3df8091aadcdb1c2fa":{"kind":"REWORK","_number":8,"created":"2020-05-11 16:52:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/8"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 16:51:57.000000000","tz":-300},"subject":"Pass full environ copy to sdist subprocess","message":"Pass full environ copy to sdist subprocess\n\nIn order to tell pbr not to build release notes and create CHANGELOG and\nAUTHORS files, we need to pass flags in as environment variables. We\ncan\u0027t just pass these flags though, as that will exclude the rest of the\nnormal environment set in. So copy the current environment, append our\nupdates, and pass that in explicitly to the subprocess call.\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04563c8ee317a89a26a46d3df8091aadcdb1c2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04563c8ee317a89a26a46d3df8091aadcdb1c2fa"}]},"branch":"refs/heads/master"},"8549c3957556959ca4044b2a4e4e5258fbeceadd":{"kind":"REWORK","_number":9,"created":"2020-05-11 18:01:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/9"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 18:00:30.000000000","tz":-300},"subject":"Pass full environ copy to sdist subprocess","message":"Pass full environ copy to sdist subprocess\n\nIn order to tell pbr not to build release notes and create CHANGELOG and\nAUTHORS files, we need to pass flags in as environment variables. We\ncan\u0027t just pass these flags though, as that will exclude the rest of the\nnormal environment set in. So copy the current environment, append our\nupdates, and pass that in explicitly to the subprocess call.\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8549c3957556959ca4044b2a4e4e5258fbeceadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8549c3957556959ca4044b2a4e4e5258fbeceadd"}]},"branch":"refs/heads/master"},"f402cde902143bfb06641f3154ad834c08218cef":{"kind":"REWORK","_number":10,"created":"2020-05-11 22:09:28.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/726838/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/726838/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/726838/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/726838/10"}}},"commit":{"parents":[{"commit":"c75a909d1ff641e0a95a54fd912766d97dc0d8b4","subject":"Drop PBR environment flags from sdist build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75a909d1ff641e0a95a54fd912766d97dc0d8b4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 12:54:37.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-11 22:09:19.000000000","tz":-300},"subject":"Pass full environ copy to sdist subprocess","message":"Pass full environ copy to sdist subprocess\n\nIn order to tell pbr not to build release notes and create CHANGELOG and\nAUTHORS files, we need to pass flags in as environment variables. We\ncan\u0027t just pass these flags though, as that will exclude the rest of the\nnormal environment set in. So copy the current environment, append our\nupdates, and pass that in explicitly to the subprocess call.\n\nChange-Id: Ia47e73b51b3216572d51399338e95bd4269a8d1d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f402cde902143bfb06641f3154ad834c08218cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f402cde902143bfb06641f3154ad834c08218cef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
