)]}'
{"id":"openstack%2Freleases~728558","triplet_id":"openstack%2Freleases~master~I6b84d7092010872283ed79be9f0c7a0c6aed9f18","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I6b84d7092010872283ed79be9f0c7a0c6aed9f18","subject":"Victoria release for ansible-role-lunasa-hsm","status":"MERGED","created":"2020-05-15 19:22:05.000000000","updated":"2020-09-17 09:34:41.000000000","submitted":"2020-09-17 09:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728558-1600335281031-079f7ff7","meta_rev_id":"284536df62e58c348c1ee0e131a2c685739b2e8f","_number":728558,"virtual_id_number":728558,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 09:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-17 08:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-09-16 21:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2020-09-17 09:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2020-09-17 09:18:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-16 21:56:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 15:37:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-17 08:40:31.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-17 09:18:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-17 09:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72ebad16fbc7b28b31392f20e45abba285aa0a7a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-15 19:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ee3999f1d56d5629729772c29abd5ad4f3b90fd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:23:23.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"74305bae2736d5f9ccaae932f11f50530423667c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-15 19:23:53.000000000","message":"Patch Set 1:\n\nThis will need to be added to a victoria deliverable folder. We can\u0027t add new deliverables to ussuri at this point.","accounts_in_message":[],"_revision_number":1},{"id":"f90be03546ee79c437d2fa85c85623a5118a2ffc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-15 19:30:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"201ada80cd54fa3cec65d1a00ee974d95a5bb251","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:31:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c7654805ca86e518d1832fbccfb49edd85a714bb","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-15 19:36:18.000000000","message":"Patch Set 2:\n\nThanks for the review, Sean.  I\u0027ve added the new victoria directory in the new patch.","accounts_in_message":[],"_revision_number":2},{"id":"e19bbaec64c6ee44d4810c904cf13381b9e4d1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:43:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0ae0fdea3b43ae910878bac0230b86 : SUCCESS in 8m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ccea5a6c53d49f0877dc020abd05e50 : FAILURE in 6m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/eba0ae848aa34612bf736ff99af84a16 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"271e9f90cd9a16994600c563f1742fe5082fb751","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-18 17:59:35.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM for victoria.\n\nI think the CI error is related to the fact that this is the first release for this project, other team member will confirm that point.\n\nAlso a side note I wondering why this project doesn\u0027t have a github mirror, I\u0027m not how work this part.","accounts_in_message":[],"_revision_number":2},{"id":"460309ec6169062c1e5b2b459a7f16dd472a44ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-18 18:03:20.000000000","message":"Patch Set 2: Code-Review-1\n\nFailure is due to the repo not being set up with a publishing job and missing bug information:\n\ndeliverables/victoria/barbican-ansible-roles.yaml: validate_bugtracker: No launchpad or storyboard project given\ndeliverables/victoria/barbican-ansible-roles.yaml: validate_release_type: openstack-infra/project-config/zuul.d/projects.yaml no release job specified for openstack/ansible-role-lunasa-hsm, one of [\u0027publish-to-pypi\u0027] needs to be included in [] or no release will be published","accounts_in_message":[],"_revision_number":2},{"id":"ddaece6eef1492f69caa8282e16181e1ddca42b9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-19 19:16:26.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"7e40e4105d7bfaac89002e28a27421107bf6206f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-19 19:36:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28714838b9aa7e0ad0fb8ee829ae7bf9a5e88769","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:38:36.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"67a5b815ca4b6a2204c23551faa23ddbab85b9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:44:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9fabe9bec34f17bf9d90502fa68caa : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deb25260bf264795b2fbd1fb329d2068 : FAILURE in 6m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/89d992c86a0a442d874b9191a144f157 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":3},{"id":"96c3c02360d67a3d06ab220d2dd50ad25433af1a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-26 14:26:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1b941d8fd4f76485d97a53bc04c763f571ee34aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 14:28:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4db76db8a150f2d6d62f6a4965d2073bcea3b880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:37:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e439108c6c4d3d9f9a3e88c9ddd826 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ce1cf93050c4baa8ba55dcde8adbdc8 : FAILURE in 7m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3bd6bddcb9254d7bab922265318fa13f : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":4},{"id":"8064d8fc5f3fad2971e7210628128344ded6837d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-26 21:26:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ed31c37f3eda59b3c321e38e8ab6b682283a947","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-26 21:33:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"49551ad59b9aaa17fd5d6753a98188b7d0a60dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 21:41:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a07106ddae847c0aab72f07989a2240 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4bc6519e07749198359b11813ed4595 : FAILURE in 5m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/208e81f3121e46bab3c201458cb15c11 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":6},{"id":"cb50e60982942c2313599fb02817a588539922b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-27 16:03:47.000000000","message":"Patch Set 6:\n\nIs there a patch to add this to governance? \n\nhttps://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L1874\n\nopenstack/releases is only for officially governed projects, so it either needs to be added as one, or tagging needs to be handled by the team directly.","accounts_in_message":[],"_revision_number":6},{"id":"82fe6f60585fae6bd408d3a8e246731f4b4379c9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-27 18:12:20.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nHi Sean,  Thanks for reviewing this patch so often!  Yes, this repository for HSM management is already listed under the Barbican team.  I talked to the folks on the OpenStack Ansible team, and their preference was that our team keep ownership because they don\u0027t have the knowledge or hardware to test this role.\n\nhttps://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L51\n\nI\u0027ll also be moving the two existing HSM roles under the barbican team in the near future.","accounts_in_message":[],"_revision_number":6},{"id":"926cf05cc28a8cf12c928b35dfe0a80056292b88","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-27 19:00:54.000000000","message":"Patch Set 6:\n\nI think I see the problem. The deliverable was defined as barbican-ansible-roles in governance, but this one is named ansible-role-lunasa-hsm. So I think the deliverable file just needs to be renamed.","accounts_in_message":[],"_revision_number":6},{"id":"9d78ad8b611f414157af0e891c4d2578ed05ca73","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-06-04 20:53:19.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"24502847087a30cabbc86439497441dbfa3f91ac","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:54:56.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"48ecec22d1756f052dab6b9e7eb5dfdaf39f0f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 21:25:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce25939ab3444d3bc9904cd9fa3fd00 : TIMED_OUT in 31m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47e5bc09709144c396e8ef264a43ad6f : FAILURE in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8809d78c6ede4fc9abc2dafa0201d24a : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":7},{"id":"f44e711f42228c0164839859cfd6f60136dd419a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 18:00:24.000000000","message":"Patch Set 7:\n\nDocs fix has gone through, but we still need the governance change to merge since the validation job doesn\u0027t work with depends-on patches. A recheck once that has merged should make the jobs happy.","accounts_in_message":[],"_revision_number":7},{"id":"6eda1d7b613f0bf45264945fe23633515257dbf1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-06-09 13:33:39.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"ec854157167413f4db7d1cfb440a36d9d4a0b5cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-12 15:08:37.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c7f101a78f08cf0001d62a11edca9860c2fee30a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 15:16:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e91a533870f4163bb9e3e1eaf9dd6cb : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/576f62d2d680422b8085375612946b65 : SUCCESS in 4m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/db175b419d164865acff03beb45b613a : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":7},{"id":"d31c7459a412b8b4125aa500f001a45154622af7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-12 15:37:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9a7928fbc7fb39b4651df7c14310a8188677a989","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-01 12:56:11.000000000","message":"Patch Set 7:\n\nAre you waiting for anything else with this Douglas?","accounts_in_message":[],"_revision_number":7},{"id":"fb7c0e8db1e34c7bd1138bfa618e5357bfb69ec2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 14:01:02.000000000","message":"Patch Set 7:\n\nShould we abandon this?","accounts_in_message":[],"_revision_number":7},{"id":"87cc2a393d982a9768154512f1ed3ff7f532ce0a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-16 21:39:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"eb56b4f29329c3dc075768d9bc0ab1015e015525","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-09-16 21:39:34.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"28e8d435b242098fcf60cad1e62f25a132095cbc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:56:11.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"fbc34f24f6d66133e7f1322f097e097d97b73996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:58:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60db9e035c2454aa8f9ebd934c53da9 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2af09f09e284ea99d03c380bbe1f1a4 : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/59d3f405650841b3be4eda04d080504f : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":8},{"id":"b567c413de4a93803db06cf02763d307e25a677c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-17 08:37:24.000000000","message":"Patch Set 8:\n\nWe should probably cut the stable/victoria branch immediately for this one in the same patch?","accounts_in_message":[],"_revision_number":8},{"id":"07d5d7edf74ffd6a26a5a22ab81c2ec65172ab31","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-17 08:40:31.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0fc3851fae40584f06df461db2997740f38c4f97","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-17 08:41:20.000000000","message":"Patch Set 8:\n\n(ignore me, stable branch can be picked up later)","accounts_in_message":[],"_revision_number":8},{"id":"135499e682bb805965b6f08fc0cc2e5f7495ebea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-17 09:18:30.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"220a272bbe72ca8b7674e3b7bc971be17b305319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:18:56.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2cb3f669d1ddc71da35a11096dcadd9a0573d0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:34:40.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd62bf0618248d2bae78be86d3aa67a : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa887a760b3b41d2af0c5f269a375b2f : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/223f8f9e62fa49eb818d420e9c164a02 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":8},{"id":"284536df62e58c348c1ee0e131a2c685739b2e8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:34:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4cabfda31bf5b28a8ee1ee0f46b5be33b2fd8c8d","revisions":{"cb8b2862c071b1d74ae5e08a1adb9a8f72a37189":{"kind":"REWORK","_number":1,"created":"2020-05-15 19:22:05.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/1"}}},"commit":{"parents":[{"commit":"a2d9378d848018bd1131c5fd759d922102aefb3d","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a2d9378d848018bd1131c5fd759d922102aefb3d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb8b2862c071b1d74ae5e08a1adb9a8f72a37189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb8b2862c071b1d74ae5e08a1adb9a8f72a37189"}]},"branch":"refs/heads/master"},"aee68c9a559c1415936922fc931d23616e45b4f0":{"kind":"REWORK","_number":2,"created":"2020-05-15 19:30:03.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/2"}}},"commit":{"parents":[{"commit":"a2d9378d848018bd1131c5fd759d922102aefb3d","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a2d9378d848018bd1131c5fd759d922102aefb3d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:29:55.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aee68c9a559c1415936922fc931d23616e45b4f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aee68c9a559c1415936922fc931d23616e45b4f0"}]},"branch":"refs/heads/master"},"ca9e8adfe5f2f5bf916784072aebdeb86fc43103":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-05-19 19:36:29.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/3"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 19:35:55.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nDepends-On: Id0d49459d84ad9540ef373a838a01ca0ca06486c\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca9e8adfe5f2f5bf916784072aebdeb86fc43103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca9e8adfe5f2f5bf916784072aebdeb86fc43103"}]},"branch":"refs/heads/master"},"c792772cd54a05cd9fc0124becf3f62ee60ec4cd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-26 14:26:13.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/4"}}},"commit":{"parents":[{"commit":"6a3549bc0bd6550f6e8f6dc391030edf848c69a2","subject":"Merge \"Release ovsddbapp 0.15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3549bc0bd6550f6e8f6dc391030edf848c69a2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-26 14:26:13.000000000","tz":0},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nDepends-On: Id0d49459d84ad9540ef373a838a01ca0ca06486c\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c792772cd54a05cd9fc0124becf3f62ee60ec4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c792772cd54a05cd9fc0124becf3f62ee60ec4cd"}]},"branch":"refs/heads/master"},"4def9e7a7194066d8380c54f776063c24789a620":{"kind":"REWORK","_number":5,"created":"2020-05-26 21:26:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/5"}}},"commit":{"parents":[{"commit":"6a3549bc0bd6550f6e8f6dc391030edf848c69a2","subject":"Merge \"Release ovsddbapp 0.15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3549bc0bd6550f6e8f6dc391030edf848c69a2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-26 21:26:50.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nDepends-On: Id0d49459d84ad9540ef373a838a01ca0ca06486c\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4def9e7a7194066d8380c54f776063c24789a620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4def9e7a7194066d8380c54f776063c24789a620"}]},"branch":"refs/heads/master"},"3fa60ac824866e422606d46b63df5f7894baf2d0":{"kind":"REWORK","_number":6,"created":"2020-05-26 21:33:23.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/6"}}},"commit":{"parents":[{"commit":"6a3549bc0bd6550f6e8f6dc391030edf848c69a2","subject":"Merge \"Release ovsddbapp 0.15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3549bc0bd6550f6e8f6dc391030edf848c69a2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-26 21:33:15.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nDepends-On: Id0d49459d84ad9540ef373a838a01ca0ca06486c\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3fa60ac824866e422606d46b63df5f7894baf2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3fa60ac824866e422606d46b63df5f7894baf2d0"}]},"branch":"refs/heads/master"},"b92cacf5fb081ead083e22c41ec76c986801fba5":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-06-04 20:53:19.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/7"}}},"commit":{"parents":[{"commit":"6a3549bc0bd6550f6e8f6dc391030edf848c69a2","subject":"Merge \"Release ovsddbapp 0.15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3549bc0bd6550f6e8f6dc391030edf848c69a2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-06-04 20:52:44.000000000","tz":-300},"subject":"Initial release for ansible-role-lunasa-hsm","message":"Initial release for ansible-role-lunasa-hsm\n\nDepends-On: I4029dabd3de8468e50da10aedda3b82b3c08e72b\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b92cacf5fb081ead083e22c41ec76c986801fba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b92cacf5fb081ead083e22c41ec76c986801fba5"}]},"branch":"refs/heads/master"},"4cabfda31bf5b28a8ee1ee0f46b5be33b2fd8c8d":{"kind":"REWORK","_number":8,"created":"2020-09-16 21:39:16.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/58/728558/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/728558/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/728558/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/728558/8"}}},"commit":{"parents":[{"commit":"d37be6ad608947302640fdb64ee1e7bdeff32128","subject":"Release final python-cinderclient for victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d37be6ad608947302640fdb64ee1e7bdeff32128"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-15 19:21:57.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-09-16 21:39:07.000000000","tz":-300},"subject":"Victoria release for ansible-role-lunasa-hsm","message":"Victoria release for ansible-role-lunasa-hsm\n\nDepends-On: I4029dabd3de8468e50da10aedda3b82b3c08e72b\nChange-Id: I6b84d7092010872283ed79be9f0c7a0c6aed9f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cabfda31bf5b28a8ee1ee0f46b5be33b2fd8c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cabfda31bf5b28a8ee1ee0f46b5be33b2fd8c8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
