)]}'
{"id":"openstack%2Freleases~729308","triplet_id":"openstack%2Freleases~master~I0dcd8146f057a9c3b53c54c714c20b7e9796254d","project":"openstack/releases","branch":"master","topic":"rm-efried","hashtags":[],"change_id":"I0dcd8146f057a9c3b53c54c714c20b7e9796254d","subject":"Remove efried as placement release liaison","status":"MERGED","created":"2020-05-19 15:52:13.000000000","updated":"2020-05-20 11:55:52.000000000","submitted":"2020-05-20 11:55:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729308-1589975752773-278ba6fa","meta_rev_id":"0be5f324b7cf8d3c6bdfd1c515d55616ef4e1ec9","_number":729308,"virtual_id_number":729308,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 11:55:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-19 15:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-20 11:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-20 11:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 15:58:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-20 11:44:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-20 11:55:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68ebaeaeb27817b3aaaee80e73fbe96ac6519f7b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-19 15:52:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9213882f88c6517814ff300a523872d78266c78","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-19 15:53:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f037e3ea77bef4322cead7b1afbc2b76f6062fde","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-19 15:53:52.000000000","message":"Topic changed from rm-mriedem to rm-efried","accounts_in_message":[],"_revision_number":2},{"id":"81f85d902334f21be988342845ae44d17941ac3a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 15:56:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08adcc44562ba5a9a88f02481b0bfab6987bec86","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-19 15:56:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0c3f6759232740c1d0513d6a28ef190a6e9132b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-19 15:57:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"09be6251bf08cac3125ce693d89c552f25673ef6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 15:58:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"62fa8f61216440105bbabb3f1d27e363b96f6e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 16:10:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25a5afe599945378540d60ac04fc9cd : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36045c74c34847efafa83268ef5170f3 : SUCCESS in 11m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed3fb88b8d104e4c9d72b2d810afe95b : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":3},{"id":"33e9afa248a90157a7f816e2586298172316449f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-20 11:44:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a190daebc40c957fcaf9d123864aa70031cc43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:44:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b1e137581738397fa8155f8a1bcbe83529c53df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:55:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0be5f324b7cf8d3c6bdfd1c515d55616ef4e1ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:55:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40e31cf4bb94db3918312f09cd91745 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8f37e66d444091b0f1fb18cb3c1809 : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f21c1187c63745fb85135d342d511132 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cb496dec1f40c1e1bf147abe20113d36556629f6","revisions":{"c979dc6dc4280ba150b21b52ecaeba82e685fb68":{"kind":"REWORK","_number":1,"created":"2020-05-19 15:52:13.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/08/729308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/729308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/729308/1"}}},"commit":{"parents":[{"commit":"15f89d85128eaacffea5445bc297863abaab6b1e","subject":"Remove mriedem as placement release liaison","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15f89d85128eaacffea5445bc297863abaab6b1e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:52:13.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:52:13.000000000","tz":0},"subject":"Remove efried as placement release liaison","message":"Remove efried as placement release liaison\n\nChange-Id: I0dcd8146f057a9c3b53c54c714c20b7e9796254d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c979dc6dc4280ba150b21b52ecaeba82e685fb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c979dc6dc4280ba150b21b52ecaeba82e685fb68"}]},"branch":"refs/heads/master"},"50fba57d066ecf05003422c34e27aa00da78c338":{"kind":"REWORK","_number":2,"created":"2020-05-19 15:53:45.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/08/729308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/729308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/729308/2"}}},"commit":{"parents":[{"commit":"15f89d85128eaacffea5445bc297863abaab6b1e","subject":"Remove mriedem as placement release liaison","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15f89d85128eaacffea5445bc297863abaab6b1e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:52:13.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:53:43.000000000","tz":0},"subject":"Remove efried as placement release liaison","message":"Remove efried as placement release liaison\n\nChange-Id: I0dcd8146f057a9c3b53c54c714c20b7e9796254d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50fba57d066ecf05003422c34e27aa00da78c338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50fba57d066ecf05003422c34e27aa00da78c338"}]},"branch":"refs/heads/master"},"cb496dec1f40c1e1bf147abe20113d36556629f6":{"kind":"REWORK","_number":3,"created":"2020-05-19 15:57:01.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/08/729308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/729308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/729308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/729308/3"}}},"commit":{"parents":[{"commit":"15f89d85128eaacffea5445bc297863abaab6b1e","subject":"Remove mriedem as placement release liaison","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15f89d85128eaacffea5445bc297863abaab6b1e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:52:13.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-05-19 15:56:59.000000000","tz":0},"subject":"Remove efried as placement release liaison","message":"Remove efried as placement release liaison\n\nChange-Id: I0dcd8146f057a9c3b53c54c714c20b7e9796254d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb496dec1f40c1e1bf147abe20113d36556629f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb496dec1f40c1e1bf147abe20113d36556629f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
