)]}'
{"id":"openstack%2Freleases~729376","triplet_id":"openstack%2Freleases~master~I9cde1aae57c34c228222d3e893e9c4eb763d69c1","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I9cde1aae57c34c228222d3e893e9c4eb763d69c1","subject":"Release ovsdbapp 1.3.0","status":"MERGED","created":"2020-05-19 19:55:20.000000000","updated":"2020-05-27 11:31:17.000000000","submitted":"2020-05-27 11:31:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729376-1590579077671-231207c7","meta_rev_id":"ec01ae0c1ef11156c371c4157a2a937318613927","_number":729376,"virtual_id_number":729376,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 11:31:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 13:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-26 08:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-05-27 02:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-05-27 11:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-26 15:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-05-27 11:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-26 00:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 07:43:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-26 08:49:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-26 13:55:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-27 02:24:39.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-27 11:18:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-27 11:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4c591497e316aea233ef23b2f205cad02938c30","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 19:55:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3ff963f52dc61d92bec31b8198bba3b27f68da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:04:22.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a871976e86984480aa256860d560dac4 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdd49121e5de4d50a9b527250a50af77 : FAILURE in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9aeb5006898c4df38838e071f176bf25 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"71ad99beaef1467fe1ed02807ccf3b13c26debf6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 20:06:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8d37f3593a6b0f0067b73102c8bff657cc201f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 20:18:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffbf52675a98f1208bc310c76c819e5c7176f87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:19:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c04cd9ad80445d986a0e554aed9510 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fab39ef54a0499ba2655b9fe2887ec1 : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0b1e24b1b2a74febad6a5d445619b3a8 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0f016e49113fe44cab4cdc0fe3b2d92988c23830","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-19 21:33:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94a9ab0120de2f622a40018df9d8f768cf153358","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 22:38:34.000000000","message":"Patch Set 2:\n\nslawek: Unless I\u0027m mistaken, that isn\u0027t the head. git log is just showing things in creation order, not commit order.","accounts_in_message":[],"_revision_number":2},{"id":"7863693da395121e5d027443f191dbfff854e1c2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 22:39:59.000000000","message":"Patch Set 2:\n\nmisread. I\u0027ve never pointed to the merge commit before. Is that really what we normally do?","accounts_in_message":[],"_revision_number":2},{"id":"9d9992ec743908028fb36b135bf790664d3c3868","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 22:41:33.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"03293fd9f56a948e577adc4c8d8baf14ada4f796","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 22:42:27.000000000","message":"Patch Set 2:\n\n\u003e misread. I\u0027ve never pointed to the merge commit before. Is that\n \u003e really what we normally do?\n\nYeah, this should really point to there.\n\nIt should be picked up if you ran:\n\ntools/new_release.sh victoria ovsdbapp feature","accounts_in_message":[],"_revision_number":2},{"id":"2a5153cac2a57b0b7f9912324b584479b9cb7392","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 22:46:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c2dcb31d06c35184181e3674f4a8a55181949df","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 22:46:23.000000000","message":"Patch Set 2:\n\ngreat, i\u0027ll use that tool in the future. :D thanks everybody.","accounts_in_message":[],"_revision_number":2},{"id":"e331269a6da2e11aa2837abe4565263b9743a92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:00:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3501a7d3f0594f65b7b4177b952e4fe3 : SUCCESS in 9m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f7784bf7c3a43a994c98603429eedf8 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b95b3cbe4aed4f888c0114d039cd33fb : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"dbf639762293e8234cbb1b88e5df0301cc46ad48","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-20 15:19:05.000000000","message":"Patch Set 3: Code-Review-1\n\nThx for clarification with hashes but now I think we also need to get this one https://review.opendev.org/#/c/729627/ to be in, otherwise we will break neutron gate which is using released version of ovsdbapp.","accounts_in_message":[],"_revision_number":3},{"id":"faaa05a74383bddfa96945b68d42d758b741a6e0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 13:52:36.000000000","message":"Patch Set 3:\n\ntest fix merged in neutron","accounts_in_message":[],"_revision_number":3},{"id":"3f5a90637ecc1784b9a3b3bc258df0c31266b527","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 20:43:00.000000000","message":"Patch Set 3:\n\nWith merged test fix, periodic ovsdbapp-master job passes: https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_99e/730116/2/check/openstack-tox-py36-with-ovsdbapp-master/99ef66f/testr_results.html","accounts_in_message":[],"_revision_number":3},{"id":"37a57ff1a0bc993e5d385791108f7efc19716591","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 20:56:51.000000000","message":"Patch Set 3:\n\n\u003e With merged test fix, periodic ovsdbapp-master job passes:\n \u003e https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_99e/730116/2/check/openstack-tox-py36-with-ovsdbapp-master/99ef66f/testr_results.html\n\nIt sounded like that fix needs to be included in the released code. Do you need to update this patch to point at a more recent commit hash to pick that up?","accounts_in_message":[],"_revision_number":3},{"id":"6dac38b92e710bc52869a805fe73840522c36c62","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 21:43:03.000000000","message":"Patch Set 3:\n\n\u003e It sounded like that fix needs to be included in the released code. Do you need to update this patch to point at a more recent commit hash to pick that up?\n\nThe merged fix was for tests in the Neutron project and not in ovsdbapp, so we should be good here.","accounts_in_message":[],"_revision_number":3},{"id":"efd524a69221385a83a2f761058a55f73eebc373","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-22 07:43:03.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"df7610f44e799f2f2d5d816c31215ebba456a0cc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 07:45:46.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"2b20756010e3303cac6fbeccebfd8ac1f16da652","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-22 15:11:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"21799d3272a0a4a9a1a6b4430b1232eab01411a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-22 18:17:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"217a40f6d0b4426b4788885d45957157fcb64f4d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-25 13:03:58.000000000","message":"Patch Set 3:\n\nIf we bump the minor version in ussuri deliverable (in https://review.opendev.org/#/c/730109/, 1.2.0 will be used there.\nIt looks better to wait for a while before +A.","accounts_in_message":[],"_revision_number":3},{"id":"97832f16e2824600927ee051d505db54eb975529","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-25 18:08:17.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"cbb2a10b6ba1ffa3329129973678b0e211d03cf6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-25 20:56:14.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"20aa676164864a64c4fd0d45e0ae1fba483be788","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 20:57:17.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"edf8fdfaaad5d1de60bc9c447cedb833fe4958cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 21:08:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c771c4714be64be48c5533fa8c56a25f : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/134e96641ec04580831c3c39c6fd6ac3 : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7aba3028595d40a09a1207f975cc73b5 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"9dff557068a5dbb437cc97bbac74dfc17bdef6fc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-26 00:24:27.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1021911c870c2cfef4318d26ef86e6876d90624b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 00:25:25.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"fbae7d2cc1be4c72a804be5cb01b2a08bd164d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 00:31:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2386ae00ffcc400884ece9d57bf78d4e : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f21fd13bfa046ffade05fc16da1ae18 : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/044a7de1b9e64e2abb9bdcf0a4ae74a1 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":5},{"id":"f40211bcc656a1150abc766b5b352bb10adc5bc0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 08:49:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b76326ac615c720719d15f45ef84ebf181a8f5ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 13:55:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b55dd240c3d644860308ad1f8c3e7a7820d1d8f9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-26 15:14:10.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"510bffa59a9b165756f8d48953eccd9d6fb10ab6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-26 15:19:10.000000000","message":"Patch Set 5:\n\nEverything looks good to me. Eagerly awaiting the w+1 :D","accounts_in_message":[],"_revision_number":5},{"id":"ded81b728d1e1d9970811717c888c0ac6b0f2909","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-27 02:24:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6d31d611f1930ac9b94ae731a36ffe59231d6a5e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:18:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8106f49c0bcf16b4f8140e0a5f08ec39fd04f106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:18:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"27c0869fc82fefa4cc9388c4f173f30af2b9dfac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:31:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ec01ae0c1ef11156c371c4157a2a937318613927","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:31:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8989f36b9e94540a4946c6eddc7365a : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f188646c73546848208c20b1ac8b3cf : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5721871bf093446a98e7aebe86a16a6e : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"47a909270de496759062cb2c515da8e2e9b20c1c","revisions":{"b6be6fcf415040eb8efcf3edfdd2ebe8ed80dedb":{"kind":"REWORK","_number":1,"created":"2020-05-19 19:55:20.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/729376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/729376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/729376/1"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"subject":"Release ovsdbapp 1.2.0","message":"Release ovsdbapp 1.2.0\n\nAdds new feature, indexed lookups.\n\nChange-Id: I9cde1aae57c34c228222d3e893e9c4eb763d69c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6be6fcf415040eb8efcf3edfdd2ebe8ed80dedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6be6fcf415040eb8efcf3edfdd2ebe8ed80dedb"}]},"branch":"refs/heads/master"},"c7feba245ed588b63c756dea2a4d43460b9c59b2":{"kind":"REWORK","_number":2,"created":"2020-05-19 20:06:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/729376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/729376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/729376/2"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 20:06:19.000000000","tz":-300},"subject":"Release ovsdbapp 1.2.0","message":"Release ovsdbapp 1.2.0\n\nAdds new feature, indexed lookups.\n\nChange-Id: I9cde1aae57c34c228222d3e893e9c4eb763d69c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7feba245ed588b63c756dea2a4d43460b9c59b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7feba245ed588b63c756dea2a4d43460b9c59b2"}]},"branch":"refs/heads/master"},"2f1983f7ed9f2f89ddadbd9d06f64df9bc2713fc":{"kind":"REWORK","_number":3,"created":"2020-05-19 22:46:00.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/729376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/729376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/729376/3"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 22:45:54.000000000","tz":-300},"subject":"Release ovsdbapp 1.2.0","message":"Release ovsdbapp 1.2.0\n\nAdds new feature, indexed lookups.\n\nChange-Id: I9cde1aae57c34c228222d3e893e9c4eb763d69c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f1983f7ed9f2f89ddadbd9d06f64df9bc2713fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f1983f7ed9f2f89ddadbd9d06f64df9bc2713fc"}]},"branch":"refs/heads/master"},"d909449286d4e4d8f511f1f71313298fbc39c438":{"kind":"REWORK","_number":4,"created":"2020-05-25 20:56:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/729376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/729376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/729376/4"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-25 20:56:11.000000000","tz":0},"subject":"Release ovsdbapp 1.2.0","message":"Release ovsdbapp 1.2.0\n\nAdds new feature, indexed lookups.\n\nChange-Id: I9cde1aae57c34c228222d3e893e9c4eb763d69c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d909449286d4e4d8f511f1f71313298fbc39c438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d909449286d4e4d8f511f1f71313298fbc39c438"}]},"branch":"refs/heads/master"},"47a909270de496759062cb2c515da8e2e9b20c1c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-26 00:24:27.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/76/729376/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/729376/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/729376/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/729376/5"}}},"commit":{"parents":[{"commit":"4b3f6040fca0016e2b9984f239d948374ede8b40","subject":"Merge \"Add extra kayobe repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3f6040fca0016e2b9984f239d948374ede8b40"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:54:42.000000000","tz":-300},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-26 00:24:23.000000000","tz":0},"subject":"Release ovsdbapp 1.3.0","message":"Release ovsdbapp 1.3.0\n\nAdds new feature, indexed lookups.\n\nChange-Id: I9cde1aae57c34c228222d3e893e9c4eb763d69c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47a909270de496759062cb2c515da8e2e9b20c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47a909270de496759062cb2c515da8e2e9b20c1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
