)]}'
{"id":"openstack%2Freleases~730973","triplet_id":"openstack%2Freleases~master~I7f477866430990f5fdcc1365ac4156a9cf0b5e58","project":"openstack/releases","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"I7f477866430990f5fdcc1365ac4156a9cf0b5e58","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 20:12:18.000000000","updated":"2020-06-01 13:04:12.000000000","submitted":"2020-06-01 13:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730973-1591016652894-d93f1f7a","meta_rev_id":"5d29c171a603ef7c3d7372d6762de231f4c49636","_number":730973,"virtual_id_number":730973,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 13:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 12:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-28 12:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-01 12:48:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 12:07:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-06-01 12:48:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 13:04:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04514dbd2bba2cb23a0f0acd98f8cc86992a03ff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 20:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31e1ddc37266ec8d037d494a064f58e553a17186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:26:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82599b45a7044c08e7990051212f938 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d511e0cb4804bc2a9d47aed03ff2e8a : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6176386cb9234fd6bfbb977797f93ddd : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49034085d0aa47c7b366283dbac5b73d : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae0d6ec546b4db28c6b945383d945f8 : SUCCESS in 9m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/301a89103b2b4f0bafdd5299eb8c43cd : SUCCESS in 6m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6b9959171f6d46ce9379aa56cc6361a0 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"04bf71dc946a2fec0f6fe146b1bfd2df06590e01","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-28 12:07:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59bfdf704287b3cd8a06a18c33b59feec97a8dfb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 12:48:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8651b1ced91776cc32ff1060ae689f9d09d43bb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 12:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5009de4792641a18b0de5a3b56a650aca1617d32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:04:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ae212adc1735db78db504c07115a532a8e5d93f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:04:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629157bd0ff4487ab3bcb3600304938d : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31858e585efd4a3587b8e44e24de9af4 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e8974fde7ad4fbdbac7d7a1fa5b55e5 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c9bb73fbcd4f0f91afda508af9bdf2 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d4aef6d44db418597cd67416aaf1ffb : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eea9bcaa71b063c5f05e03046de21b86bb07862b","revisions":{"eea9bcaa71b063c5f05e03046de21b86bb07862b":{"kind":"REWORK","_number":1,"created":"2020-05-26 20:12:18.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/730973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/730973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/730973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/730973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/730973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/730973/1"}}},"commit":{"parents":[{"commit":"f5e8f2602c005cd42991a66c30788b33f1bd119b","subject":"TripleO GA Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5e8f2602c005cd42991a66c30788b33f1bd119b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:12:10.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:12:10.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I7f477866430990f5fdcc1365ac4156a9cf0b5e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eea9bcaa71b063c5f05e03046de21b86bb07862b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eea9bcaa71b063c5f05e03046de21b86bb07862b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
