)]}'
{"id":"openstack%2Freleases~731369","triplet_id":"openstack%2Freleases~master~Ie55802a5a9d888ab1fb6fcba385a041e3626295d","project":"openstack/releases","branch":"master","topic":"magnumclient-3.1.0","hashtags":[],"change_id":"Ie55802a5a9d888ab1fb6fcba385a041e3626295d","subject":"Release python-magnumclient 3.1.0","status":"MERGED","created":"2020-05-27 22:37:17.000000000","updated":"2020-05-28 12:25:49.000000000","submitted":"2020-05-28 12:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731369-1590668749429-48d602ed","meta_rev_id":"50bd0df12e7e265109ac49b233d6f195ed670c93","_number":731369,"virtual_id_number":731369,"owner":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 12:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-28 12:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-28 07:21:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":1,"date":"2020-05-28 06:34:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-28 12:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-27 23:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 06:34:27.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-05-28 07:21:21.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-05-28 12:16:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 12:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fca714fd423042325f6b50d5045dccc410ff57ac","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-05-27 22:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9336dfecaf12703991641dfd55ff9caa6fd492c3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 22:38:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"9696b089ed6fa74e1f175430e7d5e9f200b10fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 22:50:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b0c852153c4c1599ff1c6a26d53693 : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/245a16f2f0aa47b592f40c076660c52c : FAILURE in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/804f6da1eeb04f32bc07fab6036ccc97 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"427a428ca7510a5d77eeb78340f7062aa0ed1394","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-05-27 23:22:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffb20081716bdad872a8166eea7c52031a6932eb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:23:24.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a8e0363b9b42001b3c1d02da8c4b01709282c914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:29:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d1d52ec78e4ea2a6107b5aaf0ce5bf : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8bbc792d385940078b1365d10d89b1fa : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f26e521615b64825b2f40c948da866db : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7738caefa572ace962906d4fa154d07ed169a9d4","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-28 06:34:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ac1aad96e113f00460382656d88cc9f2cf79ba79","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-05-28 07:21:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe66227609650e6b36538c0e2a3948fa9376092e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 12:16:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e4ef55605c01c55e4b080c2c79da935501996a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:16:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9864560af48fa39d3b80b736082c4fe9c439182b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:25:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d3844c96a345aca32cdf391a2ef32a : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003f26f5565c411db26f36ba3ca84f5f : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9af5ee39d8064b4c96575854f3cf7911 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9799012fdebce4782f2c32aef9f1de77e0101df8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26a9a7cb7332f77ab0bc20b59ceed47a27dd2f44","revisions":{"4a860095d137ffad6f518a417961d84cd87b607a":{"kind":"REWORK","_number":1,"created":"2020-05-27 22:37:17.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/69/731369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/731369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/731369/1"}}},"commit":{"parents":[{"commit":"0a18586488ead3000ee78b61fe1dc073193fd31b","subject":"Merge \"Add some branch prep steps to the process doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a18586488ead3000ee78b61fe1dc073193fd31b"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-27 22:32:41.000000000","tz":720},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-27 22:32:41.000000000","tz":720},"subject":"Release python-magnumclient 3.1.0","message":"Release python-magnumclient 3.1.0\n\nChange-Id: Ie55802a5a9d888ab1fb6fcba385a041e3626295d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a860095d137ffad6f518a417961d84cd87b607a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a860095d137ffad6f518a417961d84cd87b607a"}]},"branch":"refs/heads/master"},"26a9a7cb7332f77ab0bc20b59ceed47a27dd2f44":{"kind":"REWORK","_number":2,"created":"2020-05-27 23:22:14.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/69/731369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/731369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/731369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/731369/2"}}},"commit":{"parents":[{"commit":"0a18586488ead3000ee78b61fe1dc073193fd31b","subject":"Merge \"Add some branch prep steps to the process doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a18586488ead3000ee78b61fe1dc073193fd31b"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-27 22:32:41.000000000","tz":720},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-27 23:17:36.000000000","tz":720},"subject":"Release python-magnumclient 3.1.0","message":"Release python-magnumclient 3.1.0\n\nChange-Id: Ie55802a5a9d888ab1fb6fcba385a041e3626295d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a9a7cb7332f77ab0bc20b59ceed47a27dd2f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a9a7cb7332f77ab0bc20b59ceed47a27dd2f44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
