)]}'
{"id":"openstack%2Freleases~731472","triplet_id":"openstack%2Freleases~master~I845c1a4197ad2ec5cf439d7df3c129be266592fc","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I845c1a4197ad2ec5cf439d7df3c129be266592fc","subject":"pymod2pkg: release 0.24","status":"MERGED","created":"2020-05-28 08:46:59.000000000","updated":"2020-05-28 12:16:51.000000000","submitted":"2020-05-28 12:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731472-1590668211310-65991141","meta_rev_id":"377712ccd61d376b270458dd94275b301332b394","_number":731472,"virtual_id_number":731472,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 12:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2020-05-28 12:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 09:20:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2020-05-28 12:07:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-28 08:48:55.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-28 09:22:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 08:48:55.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-05-28 09:20:17.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-05-28 12:07:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-28 12:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c044754b66e84207bf2c83c4547622403537b737","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-28 08:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d8e14125abb986c4d3cd8e4b1beecc30b1ebfb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 08:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/048932c2cc884107917a9c9a56706309 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3b97c60c78e42e484012a1a13db9b5d : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d3982c0ef7454148b95da190ff42de8f : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cc32cc680c6adcffb8b3f4151bf6fbaa58a36f0c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 09:20:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e1691131cf0188372c8af94528e1bc34f1a518a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:22:21.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ef1dfee5e60ec4240b9fe8bfc87b8d7ccd47cc54","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-28 12:07:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c78eb82f0b4e6441bf774f957a3117e4c60b70c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:07:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc50cbd9100179bc05b5339a142caf9aa5c6dfce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:16:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d353bf365b8461dbdc93672fcec5ef8 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a981d19323554264bc899d99b94ffcd9 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fe266bba84c45a18f8d9cbedd11c194 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"377712ccd61d376b270458dd94275b301332b394","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:16:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23ff0ead5229cc900030237e51489dff42ebab60","revisions":{"23ff0ead5229cc900030237e51489dff42ebab60":{"kind":"REWORK","_number":1,"created":"2020-05-28 08:46:59.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/72/731472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/731472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/731472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/731472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/731472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/731472/1"}}},"commit":{"parents":[{"commit":"0a18586488ead3000ee78b61fe1dc073193fd31b","subject":"Merge \"Add some branch prep steps to the process doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a18586488ead3000ee78b61fe1dc073193fd31b"}]}],"author":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2020-05-28 08:45:33.000000000","tz":120},"committer":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2020-05-28 08:45:33.000000000","tz":120},"subject":"pymod2pkg: release 0.24","message":"pymod2pkg: release 0.24\n\nCleanup py27 support, add ansible-runner\n\nChange-Id: I845c1a4197ad2ec5cf439d7df3c129be266592fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23ff0ead5229cc900030237e51489dff42ebab60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23ff0ead5229cc900030237e51489dff42ebab60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
