)]}'
{"id":"openstack%2Freleases~732734","triplet_id":"openstack%2Freleases~master~Ib96cefe0645de1201fa06e9150a771d8ab94d2f9","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ib96cefe0645de1201fa06e9150a771d8ab94d2f9","subject":"Add note for Extended Maintenance","status":"MERGED","created":"2020-06-02 16:23:18.000000000","updated":"2020-06-03 12:15:56.000000000","submitted":"2020-06-03 12:15:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732734-1591186556027-13227909","meta_rev_id":"157b8156a34ec803e24ebd6883d65ca626fdfb50","_number":732734,"virtual_id_number":732734,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-03 09:55:34.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 12:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-03 12:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-06-03 10:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-03 12:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 10:33:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-03 12:08:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 12:15:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5520b0bba0ccf5834e12234f7c92fa03e47cd7fb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-02 16:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e8e26eda020b0b578abbf45f89506148d286a9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-02 16:25:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nNot quite happy with how it looks in the table. Maybe a link would be better? Any thought on this?","accounts_in_message":[],"_revision_number":1},{"id":"ce480aed2d1ac0e52f48c6069bb971de7fd47a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed036ec35724b26bd2675da920bda49 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d1f06dff78d35501f32d39bfe08da85cbba736f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-02 16:34:38.000000000","message":"Patch Set 1:\n\nThanks Elod. Would it be possible to just add a link to a note below the table to explain this rather than repeating the text on each instance?","accounts_in_message":[],"_revision_number":1},{"id":"90f1ef313519207f2c2d733649388779526dc3e3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-02 16:43:07.000000000","message":"Patch Set 1:\n\n\u003e Thanks Elod. Would it be possible to just add a link to a note\n \u003e below the table to explain this rather than repeating the text on\n \u003e each instance?\n\nYes, that sounds better. Let me experiment with that. :)","accounts_in_message":[],"_revision_number":1},{"id":"dcc6250fbebc56a18447ac434cf8538e56fe5463","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-03 09:44:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a19ec7ba4bbabbc2b7842f25dd6e6732950897c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 09:53:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c10f3c9b21f4fda9680aceb4e370af2 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":2},{"id":"64e0793d00025a7896cfce82be0d0831393335b8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-03 09:55:34.000000000","message":"Patch Set 2:\n\nNow it looks maybe a bit better, see generated doc:\n\nhttps://58a693ff35d61e224506-11d92d8452a346ca231ad13d26a55a7d.ssl.cf1.rackcdn.com/732734/2/check/openstack-tox-docs/5c10f3c/docs/\n\nLet me know if there\u0027s any typo or something needs to be rephrased :)","accounts_in_message":[],"_revision_number":2},{"id":"6de675c1793dd7eaa39734d9b3ca14ae9c390f00","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 10:33:39.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"9a2e41f1d8debb8e5dceedbeb574492e3ea8d605","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:08:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks great, thanks Elod!","accounts_in_message":[],"_revision_number":2},{"id":"fb8112640eaa3479ede28ce599ebbc956b3d0886","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:08:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aaac590a383b64093bb93e46428d510047c3e056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:15:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cefcd9c3a21446db7b32a98a036eef1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd31ae6e86843d8a3069476663148f5 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":2},{"id":"157b8156a34ec803e24ebd6883d65ca626fdfb50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:15:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c48a75fb908c315bc382ee3675ff752c380c1a1e","revisions":{"1d0ba320a7fcbdddeef4dcba0207dcd690cc2e9e":{"kind":"REWORK","_number":1,"created":"2020-06-02 16:23:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/34/732734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/732734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/732734/1"}}},"commit":{"parents":[{"commit":"9ae1d5a25a43697f5ef0c5b6b520ca7f06cf985d","subject":"Merge \"Set ocata unmaintained transition date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ae1d5a25a43697f5ef0c5b6b520ca7f06cf985d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-06-02 12:01:44.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-06-02 16:22:21.000000000","tz":120},"subject":"Add note for Extended Maintenance","message":"Add note for Extended Maintenance\n\nExtended Maintenance does not mean by default that every project is\nmaintained. It could happen that a project is not actively maintained\nor unmaintained or even it\u0027s team marked it as End of Life. This patch\nadds a note to clarify that.\n\nChange-Id: Ib96cefe0645de1201fa06e9150a771d8ab94d2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d0ba320a7fcbdddeef4dcba0207dcd690cc2e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d0ba320a7fcbdddeef4dcba0207dcd690cc2e9e"}]},"branch":"refs/heads/master"},"c48a75fb908c315bc382ee3675ff752c380c1a1e":{"kind":"REWORK","_number":2,"created":"2020-06-03 09:44:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/34/732734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/732734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/732734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/732734/2"}}},"commit":{"parents":[{"commit":"9ae1d5a25a43697f5ef0c5b6b520ca7f06cf985d","subject":"Merge \"Set ocata unmaintained transition date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ae1d5a25a43697f5ef0c5b6b520ca7f06cf985d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-06-02 12:01:44.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-06-03 09:42:10.000000000","tz":120},"subject":"Add note for Extended Maintenance","message":"Add note for Extended Maintenance\n\nExtended Maintenance does not mean by default that every project is\nmaintained. It could happen that a project is not actively maintained\nor unmaintained or even it\u0027s team marked it as End of Life. This patch\nadds a note to clarify that.\n\nAlso reformat a previous note.\n\nChange-Id: Ib96cefe0645de1201fa06e9150a771d8ab94d2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c48a75fb908c315bc382ee3675ff752c380c1a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c48a75fb908c315bc382ee3675ff752c380c1a1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
