)]}'
{"id":"openstack%2Freleases~733188","triplet_id":"openstack%2Freleases~master~I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82","project":"openstack/releases","branch":"master","topic":"oslo-master","hashtags":[],"change_id":"I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82","subject":"Oslo releases for 2020-06-03","status":"MERGED","created":"2020-06-03 15:55:19.000000000","updated":"2020-06-04 10:28:33.000000000","submitted":"2020-06-04 10:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733188-1591266513640-9d8ebeac","meta_rev_id":"9e21927e6f3c5fc0a47e2772a1fbfe83d37e7880","_number":733188,"virtual_id_number":733188,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-03 21:21:47.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 10:28:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-06-04 10:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-06-04 07:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-06-04 10:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-06-03 21:21:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 07:30:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-04 10:12:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-06-04 10:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eeb1cd36e6039155ff6c02c9daa9bf68108eeed4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-03 15:55:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31fbcf7c1ba0ef7e390ec50ddd3ec8c6fa36c32e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:56:22.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c7a500cdc56629d758b3521474c05320c98b781b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:06:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7904c037682a446e91ce960d8f5498b8 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6084b278857d43d2b4ae9ceba09a4aff : SUCCESS in 7m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a3f932c252ad4deda210dd39791d8cd2 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1e41ac16f313cb34cf44d8db5bcfc01ed6d4ecf0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 16:52:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nHello Ben,\n\nPlease see my inline comments, thanks.","accounts_in_message":[],"_revision_number":1},{"id":"eb6e30af6291e65105d83bbcd6d6227bb3e821f9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-03 19:19:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"843248ace3baa09137c1b32d21ab61ac6119c387","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 20:06:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3e761f99138f0ef631412417a0a44077e922ee3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-03 21:20:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a3c2b9a0bcdb1f813b026ac4a5a59cce370db32","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 21:21:41.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1d6990c8ca1edd83eb56329de1ecd4c82030829a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-03 21:21:47.000000000","message":"Patch Set 2:\n\nFixed the stevedore version.","accounts_in_message":[],"_revision_number":2},{"id":"dc3e4e90d1852e7b66e6453cf87a62ada332d7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 21:29:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e755f288fdee4c3f849f24d3634c421a : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad3d3120ab40417c9c0a2317a05c91d3 : SUCCESS in 8m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/933b10d7d5674917b455d37824b659c0 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7cc293aba70dcc2784fc2240b34fe848dee29342","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 07:30:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"307be45ef8db1d3cc0c83e8b0c93844d55f34c60","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-04 10:12:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bdb06aadfb6d05a5530ce3e9a95c63f94647930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:12:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78d51bfb99a9eb9e57ddb4533d91720b2c9b5cac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:28:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cdf7cdb8c4a06a46be1875570aea02139dc10022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:28:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf94f72465d74467b5bb195d22e34f27 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1760e4b8cb944808f9f990ae0ed4dec : SUCCESS in 10m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e5b6b07ffb540f6ad83ebb9af1f5b75 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4acac3e5e47537a87f2796d9f6f40df6a04b062","revisions":{"64aa0dfc5fe8f56d69e00fa5f387756157015520":{"kind":"REWORK","_number":1,"created":"2020-06-03 15:55:19.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/88/733188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/733188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/733188/1"}}},"commit":{"parents":[{"commit":"2f2ed9fc74365d39f8ba80233af57c33aced96e3","subject":"Merge \"Release python-troveclient 4.0.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f2ed9fc74365d39f8ba80233af57c33aced96e3"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-06-03 15:53:43.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-06-03 15:53:43.000000000","tz":0},"subject":"Oslo releases for 2020-06-03","message":"Oslo releases for 2020-06-03\n\nI\u0027m not sure all of these strictly need to be released, but some of\nthe projects had quite a few patches piling up which makes it harder\nto tell at a glance whether there are release-worthy patches. Things\nlike the doc requirements changes may be helpful for packagers too.\n\nChange-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64aa0dfc5fe8f56d69e00fa5f387756157015520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64aa0dfc5fe8f56d69e00fa5f387756157015520"}]},"branch":"refs/heads/master"},"f4acac3e5e47537a87f2796d9f6f40df6a04b062":{"kind":"REWORK","_number":2,"created":"2020-06-03 21:20:24.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/88/733188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/733188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/733188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/733188/2"}}},"commit":{"parents":[{"commit":"2f2ed9fc74365d39f8ba80233af57c33aced96e3","subject":"Merge \"Release python-troveclient 4.0.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f2ed9fc74365d39f8ba80233af57c33aced96e3"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-06-03 15:53:43.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-06-03 19:20:10.000000000","tz":0},"subject":"Oslo releases for 2020-06-03","message":"Oslo releases for 2020-06-03\n\nI\u0027m not sure all of these strictly need to be released, but some of\nthe projects had quite a few patches piling up which makes it harder\nto tell at a glance whether there are release-worthy patches. Things\nlike the doc requirements changes may be helpful for packagers too.\n\nChange-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4acac3e5e47537a87f2796d9f6f40df6a04b062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4acac3e5e47537a87f2796d9f6f40df6a04b062"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
