)]}'
{"id":"openstack%2Freleases~734940","triplet_id":"openstack%2Freleases~master~Ic55ffdd28a10ce2cd6a3c51f7ade077c7370171c","project":"openstack/releases","branch":"master","topic":"train-stable","hashtags":[],"change_id":"Ic55ffdd28a10ce2cd6a3c51f7ade077c7370171c","subject":"Release masakari-monitors for stable/train","status":"MERGED","created":"2020-06-10 21:13:36.000000000","updated":"2020-08-19 11:35:41.000000000","submitted":"2020-08-19 11:35:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734940-1597836941890-0b925b31","meta_rev_id":"c8d836aa39184d6e92712fb0f5e7d50087616518","_number":734940,"virtual_id_number":734940,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-19 11:35:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-19 11:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-19 08:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-08-17 19:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-08-19 04:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-19 11:25:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-19 04:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 19:28:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-19 04:22:09.000000000","updated_by":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"reviewer":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"state":"REVIEWER"},{"updated":"2020-08-19 08:49:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-19 11:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adbfc37ed3a0534e8b46dd106e9439de31945137","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 21:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bdaf8d50559533ed9bec6a454e9eecd6256fde0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 21:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ffada6dc824891b198ed60c3d425a6 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/110de72dc35646a89b2b0fe102b22a4c : SUCCESS in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4092e0e8eb914d45bf763fe676c1fbe2 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"05268ddf2568e24128868cd9807b51e874c44fc5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 18:25:14.000000000","message":"Abandoned\n\nNo response from team, and this was just proposed as a way to help make sure stable commits were actually being released. So abandoning this patch, but feel free to restore at any point, or propose a new release request, when ready for a stable release.","accounts_in_message":[],"_revision_number":1},{"id":"df1e391729fd3cd6f7c9a5e14cac276c4ecbdc07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 18:22:50.000000000","message":"Patch Set 1:\n\nI\u0027d like to restore it; should I repropose then?","accounts_in_message":[],"_revision_number":1},{"id":"4c4a629b848749af7229cbdac82ded2b05e2b0eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 18:26:56.000000000","message":"Patch Set 1:\n\n\u003e I\u0027d like to restore it; should I repropose then?\n\nSure, if the commit hash hasn\u0027t changed, you can just restore and leave a +1. Otherwise, feel free to update with the current commit to use.","accounts_in_message":[],"_revision_number":1},{"id":"98eef43de1e3303f473beebc72177567236ac558","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 18:45:33.000000000","message":"Patch Set 1:\n\n\u003e \u003e I\u0027d like to restore it; should I repropose then?\n \u003e \n \u003e Sure, if the commit hash hasn\u0027t changed, you can just restore and\n \u003e leave a +1. Otherwise, feel free to update with the current commit\n \u003e to use.\n\nSure thing; the commit hash has not changed - I just can\u0027t find the button to do the restore. I guess it requires being the change owner and/or core. I\u0027ll leave a +1 if you restore it.","accounts_in_message":[],"_revision_number":1},{"id":"1749d1bc4530fde47195a1de75c48426bedac4b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 19:05:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e9d9abd3e1f9fb561b535909074570a85973ea95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 19:19:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c6d4ec6833483e9a7b559a1037d21a : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3adfbb02ad884b8d97f69a36fe5fa3a5 : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0107363ed5a3405f8f6e5816c3499203 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1255e306d2188815afd0a384a724870c5b52fdb5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 19:28:01.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks, Sean","accounts_in_message":[],"_revision_number":1},{"id":"e5eb842529da8c31e4cb35950b804bd17e17db7f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 19:28:25.000000000","message":"Patch Set 1:\n\nthis makes Train usable with pacemaker-remotes","accounts_in_message":[],"_revision_number":1},{"id":"df7f28f5342e3b2a6b5a8df34f1fc6f91a0eec8d","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2020-08-19 04:22:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa106c54affe6a21f392e822b1e7c8b85ee492cc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 04:23:01.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"db74c95b345657681c59e588b6096d9b13c8d614","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 08:33:18.000000000","message":"Patch Set 1:\n\nHello,\n\nI would prefer to release a minor (8.1.0) version here, ussuri\u0027s cycle starting with a version 9.0.0 and b02c6b69 introduced a new public option [1].\n\nmaskari-monitors don\u0027t seem to follow the stable policy [2][3] so I think we can go ahead with a minor version. It could help us to highlight this new option.\n\nThoughts?\n\n[1] https://github.com/openstack/masakari-monitors/commit/b02c6b6931c0256f4ce6d7167c97ebb849ff3453\n[2] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n[3] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L884","accounts_in_message":[],"_revision_number":1},{"id":"e50cc74d357c72847137e4eb2b0831e2c6db9d99","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 08:38:24.000000000","message":"Patch Set 1:\n\n\u003e Hello,\n \u003e \n \u003e I would prefer to release a minor (8.1.0) version here, ussuri\u0027s\n \u003e cycle starting with a version 9.0.0 and b02c6b69 introduced a new\n \u003e public option [1].\n \u003e \n \u003e maskari-monitors don\u0027t seem to follow the stable policy [2][3] so I\n \u003e think we can go ahead with a minor version. It could help us to\n \u003e highlight this new option.\n \u003e \n \u003e Thoughts?\n \u003e \n \u003e [1] https://github.com/openstack/masakari-monitors/commit/b02c6b6931c0256f4ce6d7167c97ebb849ff3453\n \u003e [2] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n \u003e [3] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L884\n\nIt was a poor-man\u0027s fix to handle remotes properly, hence why it introduced the variable. I would not stress it this much though, either approach is fine by me. If minor bump is better, then be it a minor bump. Let\u0027s just make the release happen.","accounts_in_message":[],"_revision_number":1},{"id":"fcbd1ad9f4afecf75080f0f675ae1bc6d1d78e7e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 08:49:05.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e \u003e Hello,\n \u003e \u003e\n \u003e \u003e I would prefer to release a minor (8.1.0) version here, ussuri\u0027s\n \u003e \u003e cycle starting with a version 9.0.0 and b02c6b69 introduced a new\n \u003e \u003e public option [1].\n \u003e \u003e\n \u003e \u003e maskari-monitors don\u0027t seem to follow the stable policy [2][3] so\n \u003e I\n \u003e \u003e think we can go ahead with a minor version. It could help us to\n \u003e \u003e highlight this new option.\n \u003e \u003e\n \u003e \u003e Thoughts?\n \u003e \u003e\n \u003e \u003e [1] https://github.com/openstack/masakari-monitors/commit/b02c6b6931c0256f4ce6d7167c97ebb849ff3453\n \u003e \u003e [2] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n \u003e \u003e [3] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L884\n \u003e \n \u003e It was a poor-man\u0027s fix to handle remotes properly, hence why it\n \u003e introduced the variable. I would not stress it this much though,\n \u003e either approach is fine by me. If minor bump is better, then be it\n \u003e a minor bump. Let\u0027s just make the release happen.\n\nOk thanks for the details, so as you want. So release a bugfix version isn\u0027t an earthquake, so for this reason I +2\u0027ing to lets Sean and other core reviewers validate if it look fine for them and to lets choose to bump or not.","accounts_in_message":[],"_revision_number":1},{"id":"ce7904a4a7035ad87357be44cd72cb8010dd1771","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 09:19:40.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e Hello,\n \u003e \u003e \u003e\n \u003e \u003e \u003e I would prefer to release a minor (8.1.0) version here,\n \u003e ussuri\u0027s\n \u003e \u003e \u003e cycle starting with a version 9.0.0 and b02c6b69 introduced a\n \u003e new\n \u003e \u003e \u003e public option [1].\n \u003e \u003e \u003e\n \u003e \u003e \u003e maskari-monitors don\u0027t seem to follow the stable policy [2][3]\n \u003e so\n \u003e \u003e I\n \u003e \u003e \u003e think we can go ahead with a minor version. It could help us to\n \u003e \u003e \u003e highlight this new option.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Thoughts?\n \u003e \u003e \u003e\n \u003e \u003e \u003e [1] https://github.com/openstack/masakari-monitors/commit/b02c6b6931c0256f4ce6d7167c97ebb849ff3453\n \u003e \u003e \u003e [2] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n \u003e \u003e \u003e [3] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L884\n \u003e \u003e\n \u003e \u003e It was a poor-man\u0027s fix to handle remotes properly, hence why it\n \u003e \u003e introduced the variable. I would not stress it this much though,\n \u003e \u003e either approach is fine by me. If minor bump is better, then be\n \u003e it\n \u003e \u003e a minor bump. Let\u0027s just make the release happen.\n \u003e \n \u003e Ok thanks for the details, so as you want. So release a bugfix\n \u003e version isn\u0027t an earthquake, so for this reason I +2\u0027ing to lets\n \u003e Sean and other core reviewers validate if it look fine for them and\n \u003e to lets choose to bump or not.\n\nThanks, Hervé!","accounts_in_message":[],"_revision_number":1},{"id":"83f629c31564d2ff836ad483ef66a77d9109f6d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-19 11:25:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"083082762179c060d9d69eb5c69eed9f04b2249c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 11:25:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d61d375f8819e4eda0a08ec7950c222449fa341","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 11:35:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1a148e1844680116579affadca5dad8801e5fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 11:35:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5cc0fedbc144ec88fdc3142e7ec871 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a1dacb70614d41a2c9d86fd52d47be : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cf39f4cbaa84779aa5e65d7e5a1b63e : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1952c1a4cab8dfec75538cc9c9b5cd118e24521","revisions":{"c1952c1a4cab8dfec75538cc9c9b5cd118e24521":{"kind":"REWORK","_number":1,"created":"2020-06-10 21:13:36.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/40/734940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/734940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/734940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/734940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/734940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/734940/1"}}},"commit":{"parents":[{"commit":"855407c88842a02a9183f6256cb2227d68ce6455","subject":"Merge \"Release 3.2.0 of cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/855407c88842a02a9183f6256cb2227d68ce6455"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:13:35.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:13:35.000000000","tz":-300},"subject":"Release masakari-monitors for stable/train","message":"Release masakari-monitors for stable/train\n\nThis release picks up new commits to masakari-monitors since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nChange-Id: Ic55ffdd28a10ce2cd6a3c51f7ade077c7370171c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1952c1a4cab8dfec75538cc9c9b5cd118e24521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1952c1a4cab8dfec75538cc9c9b5cd118e24521"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
