)]}'
{"id":"openstack%2Freleases~734959","triplet_id":"openstack%2Freleases~master~I633c3d7ce3dc37080c12e1d39b8fe4afc936fa9f","project":"openstack/releases","branch":"master","topic":"train-stable","hashtags":[],"change_id":"I633c3d7ce3dc37080c12e1d39b8fe4afc936fa9f","subject":"Release puppet-aodh for stable/train","status":"ABANDONED","created":"2020-06-10 21:32:29.000000000","updated":"2020-06-18 18:33:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aee2e4b5878b176547f926cd25d177bc45f5d4eb","_number":734959,"virtual_id_number":734959,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-10 21:41:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"date":"2020-06-10 22:02:57.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 21:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 22:02:57.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"7adc1eb9544ad5dabe3fa12f4b699a854358d9a3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 21:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d74697c21c3a79971e4094e03011bc09d97daac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 21:41:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b17ba52af1f404e84fb7e040000dfbe : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e13d2a7df28842df870b0e455046fe5b : FAILURE in 6m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/38d6bf7ec2824768b0bde6a10f8904bf : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"053ba7b41ed1ac63e730eb250554fef08c8e1af1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 18:33:59.000000000","message":"Abandoned\n\nNo response from team, and this was just proposed as a way to help make sure stable commits were actually being released. So abandoning this patch, but feel free to restore at any point, or propose a new release request, when ready for a stable release.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d90db4f5baa2574a0bc10989de8ae860f0a3a16","revisions":{"9d90db4f5baa2574a0bc10989de8ae860f0a3a16":{"kind":"REWORK","_number":1,"created":"2020-06-10 21:32:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/734959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/734959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/734959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/734959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/734959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/734959/1"}}},"commit":{"parents":[{"commit":"855407c88842a02a9183f6256cb2227d68ce6455","subject":"Merge \"Release 3.2.0 of cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/855407c88842a02a9183f6256cb2227d68ce6455"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:32:28.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:32:28.000000000","tz":-300},"subject":"Release puppet-aodh for stable/train","message":"Release puppet-aodh for stable/train\n\nThis release picks up new commits to puppet-aodh since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nChange-Id: I633c3d7ce3dc37080c12e1d39b8fe4afc936fa9f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d90db4f5baa2574a0bc10989de8ae860f0a3a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d90db4f5baa2574a0bc10989de8ae860f0a3a16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
