)]}'
{"id":"openstack%2Freleases~734964","triplet_id":"openstack%2Freleases~master~I536cda70d872acc372c0bb59da677ddeebcb98f5","project":"openstack/releases","branch":"master","topic":"train-stable","hashtags":[],"change_id":"I536cda70d872acc372c0bb59da677ddeebcb98f5","subject":"Release puppet-congress for stable/train","status":"ABANDONED","created":"2020-06-10 21:33:56.000000000","updated":"2020-06-18 18:36:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"78f878ee1952f50d02c1ea6c4929825ae99748c7","_number":734964,"virtual_id_number":734964,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-10 21:46:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"date":"2020-06-10 22:02:36.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 21:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 22:02:36.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"ef41daf78b8f475245c254e171c74c15f331c929","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 21:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"825a1a3af78433ee4c11b62443d66efd9f53592f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 21:46:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e29f10b08b4374980b54c8b31ead21 : SUCCESS in 7m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e614df2299945a49c7086cb12755e21 : FAILURE in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/38e166bbb6a04ab7bbc6b2a2e605adc6 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c2c36dd96859c55f77888f8c75dc17b2fd436791","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 18:36:29.000000000","message":"Abandoned\n\nNo response from team, and this was just proposed as a way to help make sure stable commits were actually being released. So abandoning this patch, but feel free to restore at any point, or propose a new release request, when ready for a stable release.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5d2fbca01a1f6c51ea12693091c340c3ba8cc92","revisions":{"f5d2fbca01a1f6c51ea12693091c340c3ba8cc92":{"kind":"REWORK","_number":1,"created":"2020-06-10 21:33:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/64/734964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/734964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/734964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/734964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/734964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/734964/1"}}},"commit":{"parents":[{"commit":"855407c88842a02a9183f6256cb2227d68ce6455","subject":"Merge \"Release 3.2.0 of cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/855407c88842a02a9183f6256cb2227d68ce6455"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:33:54.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-10 21:33:54.000000000","tz":-300},"subject":"Release puppet-congress for stable/train","message":"Release puppet-congress for stable/train\n\nThis release picks up new commits to puppet-congress since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nChange-Id: I536cda70d872acc372c0bb59da677ddeebcb98f5\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5d2fbca01a1f6c51ea12693091c340c3ba8cc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5d2fbca01a1f6c51ea12693091c340c3ba8cc92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
