)]}'
{"id":"openstack%2Freleases~735330","triplet_id":"openstack%2Freleases~master~Ic99b469cb15e87ea9b6e3505d2280d45cd224348","project":"openstack/releases","branch":"master","topic":"cleanup-independent","hashtags":[],"change_id":"Ic99b469cb15e87ea9b6e3505d2280d45cd224348","subject":"Cleanup old independent releases","status":"MERGED","created":"2020-06-12 12:22:01.000000000","updated":"2020-06-16 09:31:18.000000000","submitted":"2020-06-16 09:31:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735330-1592299878860-a36c32e4","meta_rev_id":"1913d67ed2645d164d66a6ca113996aedaeeaa60","_number":735330,"virtual_id_number":735330,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-16 09:25:25.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 09:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-15 15:06:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":2,"date":"2020-06-16 09:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":1,"date":"2020-06-16 09:16:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-06-12 13:20:08.000000000","_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2020-06-12 13:20:08.000000000","_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 13:20:08.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2020-06-12 13:20:08.000000000","updated_by":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2020-06-15 15:06:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-16 09:16:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-16 09:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d236d1f86e8bfd1a648e46d2156defe861e39f98","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-12 12:22:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3927fd0f13687c8b15fdd5cef264e2cae636e27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 12:31:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d8f96681c644f48568aa4b742cccf8 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3c263c767764c29836b17ba34400905 : SUCCESS in 4m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4696f5b31ada4850b029a5c20a6f7407 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d4c3f5bb0da3e562841d12ebac77cc4283de196a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-12 12:45:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e77e78b93022375f57255376819395fe6005f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 12:52:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7ec5f6e39641ae8f3b310e08cb7157 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb0b8bb95dbb457fa2dd8351527f17fc : SUCCESS in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5817916cff6f4b4981483c4e08e6f003 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e4690e06a9279527be9881fcea51afdfd968da03","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-12 12:55:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"102d41cd92766dbc3278bfd0753c6f7e77f380df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 13:02:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50deebb7534d45ef8a57afc6554b8cb6 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6a1e262fdb140f782698e6347253fc2 : SUCCESS in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/56977c50e5c24d749e9f825757add482 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"fa0dcd84015548c13da03d1b889eac95bf10c6ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 15:06:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b4cd0168baf1e6f41e666c530a6022344ab8e6fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-16 09:16:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"0b37bf8b18bb1c33e60ea075b69c1a55190ebfea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:16:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab3c8f9d703beb72d12f457bf8cd093945307fe1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-06-16 09:25:25.000000000","message":"Patch Set 3:\n\nNB: the only painful one is openstack-ansible, as it started independent, was following cycle in Liberty, then back to independent during Mitaka, then following cycle again starting in Newton. But since this is all pre-Newton history I think we should still clean up... Otherwise people find https://releases.openstack.org/independent.html#openstack-ansible and assume 13.3.18 is the latest available.","accounts_in_message":[],"_revision_number":3},{"id":"e48d1cefa85d0f34d514d762e286cec56f4d0399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:31:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c6c9d2c1e4683c991f1fe7bf424e55c480155810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:31:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2424c2a290ea4bcdbb96db742b1e7a09 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04ca21160794d4c9ded5543224668a4 : SUCCESS in 10m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/601cb07e80f6484e8e37144fd401bcca : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4302776523e915edc3477ffc8621ab7931d2f5c2","revisions":{"bf21fa5f300aa40864bc34fddae00d0d7dc96210":{"kind":"REWORK","_number":1,"created":"2020-06-12 12:22:01.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/30/735330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/735330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/735330/1"}}},"commit":{"parents":[{"commit":"0092e712a63a03496c9d85e6c2af9ef0254f90df","subject":"Merge \"Octavia: EOL Ocata and Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0092e712a63a03496c9d85e6c2af9ef0254f90df"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 11:57:11.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 12:21:21.000000000","tz":120},"subject":"Cleanup old independent releases","message":"Cleanup old independent releases\n\nA lot of OpenStack deliverables start their existence as cycle-independent,\nas it is what our docs still recommend for initial releases. Most of them\nquickly turn into cycle-following deliverables, however their early\nlives as independent deliverables produce a confusing trail at [1].\n\nThis page seems to imply (for example) that Cyborg is a\ncycle-independent deliverable and that its most recent version is 0.2.0.\nThis can be very confusing, and displaying long-forgotten early releases\non releases.o.o is not worth it.\n\nNote: cycle-following deliverables that become independent actually do\nnot generate the same confusion. They stop being listed on the\ncoordinated release pages and their listing on [1] is accurate. It\u0027s the\nindependent deliverables that become tied to the cycle that result in\nconfusing information.\n\nThis change removes deliverable files for deliverables that had a couple\nof independent releases at the start of their lives but are now\nfollowing the release cycle. Those releases can still be found on\ntarballs.o.o and their tags still exist in their history. We just stop\ndisplaying them on releases.o.o.\n\nThis allows the \"Series-Independent Projects\" page[1] to only display\ndeliverables that actually *are* released in a cycle-independent manner.\n\n[1] https://releases.openstack.org/independent.html\n\nChange-Id: Ic99b469cb15e87ea9b6e3505d2280d45cd224348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf21fa5f300aa40864bc34fddae00d0d7dc96210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf21fa5f300aa40864bc34fddae00d0d7dc96210"}]},"branch":"refs/heads/master"},"b7e34099a4fa5e183566108d73a4ec2dad5dad02":{"kind":"REWORK","_number":2,"created":"2020-06-12 12:45:13.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/30/735330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/735330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/735330/2"}}},"commit":{"parents":[{"commit":"0092e712a63a03496c9d85e6c2af9ef0254f90df","subject":"Merge \"Octavia: EOL Ocata and Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0092e712a63a03496c9d85e6c2af9ef0254f90df"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 11:57:11.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 12:45:02.000000000","tz":120},"subject":"Cleanup old independent releases","message":"Cleanup old independent releases\n\nA lot of OpenStack deliverables start their existence as cycle-independent,\nas it is what our docs still recommend for initial releases. Most of them\nquickly turn into cycle-following deliverables, however their early\nlives as independent deliverables produce a confusing trail at [1].\n\nThis page seems to imply (for example) that Cyborg is a\ncycle-independent deliverable and that its most recent version is 0.2.0.\nThis can be very confusing, and displaying long-forgotten early releases\non releases.o.o is not worth it.\n\nNote: cycle-following deliverables that become independent actually do\nnot generate the same confusion. They stop being listed on the\ncoordinated release pages and their listing on [1] is accurate. It\u0027s the\nindependent deliverables that become tied to the cycle that result in\nconfusing information.\n\nThis change removes deliverable files for deliverables that had a couple\nof independent releases at the start of their lives but are now\nfollowing the release cycle. Those releases can still be found on\ntarballs.o.o and their tags still exist in their history. We just stop\ndisplaying them on releases.o.o.\n\nThis allows the \"Series-Independent Projects\" page[1] to only display\ndeliverables that actually *are* released in a cycle-independent manner.\n\n[1] https://releases.openstack.org/independent.html\n\nChange-Id: Ic99b469cb15e87ea9b6e3505d2280d45cd224348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7e34099a4fa5e183566108d73a4ec2dad5dad02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7e34099a4fa5e183566108d73a4ec2dad5dad02"}]},"branch":"refs/heads/master"},"4302776523e915edc3477ffc8621ab7931d2f5c2":{"kind":"REWORK","_number":3,"created":"2020-06-12 12:55:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/30/735330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/735330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/735330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/735330/3"}}},"commit":{"parents":[{"commit":"0092e712a63a03496c9d85e6c2af9ef0254f90df","subject":"Merge \"Octavia: EOL Ocata and Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0092e712a63a03496c9d85e6c2af9ef0254f90df"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 11:57:11.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-06-12 12:55:33.000000000","tz":120},"subject":"Cleanup old independent releases","message":"Cleanup old independent releases\n\nA lot of OpenStack deliverables start their existence as cycle-independent,\nas it is what our docs still recommend for initial releases. Most of them\nquickly turn into cycle-following deliverables, however their early\nlives as independent deliverables produce a confusing trail at [1].\n\nThis page seems to imply (for example) that Cyborg is a\ncycle-independent deliverable and that its most recent version is 0.2.0.\nThis can be very confusing, and displaying long-forgotten early releases\non releases.o.o is not worth it.\n\nNote: cycle-following deliverables that become independent actually do\nnot generate the same confusion. They stop being listed on the\ncoordinated release pages and their listing on [1] is accurate. It\u0027s the\nindependent deliverables that become tied to the cycle that result in\nconfusing information.\n\nThis change removes deliverable files for deliverables that had a couple\nof independent releases at the start of their lives but are now\nfollowing the release cycle. Those releases can still be found on\ntarballs.o.o and their tags still exist in their history. We just stop\ndisplaying them on releases.o.o.\n\nThis allows the \"Series-Independent Projects\" page[1] to only display\ndeliverables that actually *are* released in a cycle-independent manner.\n\n[1] https://releases.openstack.org/independent.html\n\nChange-Id: Ic99b469cb15e87ea9b6e3505d2280d45cd224348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4302776523e915edc3477ffc8621ab7931d2f5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4302776523e915edc3477ffc8621ab7931d2f5c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
