)]}'
{"id":"openstack%2Freleases~735440","triplet_id":"openstack%2Freleases~master~I9fd1426d315e77f2dcdc5cdeeb71e35fe7781f6b","project":"openstack/releases","branch":"master","topic":"cliff_3_3_0","hashtags":[],"change_id":"I9fd1426d315e77f2dcdc5cdeeb71e35fe7781f6b","subject":"Release 3.3.0 of cliff","status":"MERGED","created":"2020-06-13 18:30:01.000000000","updated":"2020-06-14 10:32:41.000000000","submitted":"2020-06-14 10:32:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735440-1592130761265-1d1e633c","meta_rev_id":"056ad3b26953fb3ebd82cd92919cdd2434418302","_number":735440,"virtual_id_number":735440,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-14 10:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-13 18:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-06-14 09:24:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-14 10:22:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-13 19:20:12.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-06-13 18:30:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-06-14 09:26:02.000000000","_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"date":"2020-06-13 18:30:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 18:30:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-13 18:30:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-06-13 19:20:12.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-14 09:24:32.000000000","updated_by":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"reviewer":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"state":"REVIEWER"},{"updated":"2020-06-14 09:26:02.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-06-14 10:22:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-14 10:32:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9e235cee3d695ec5630b8cc1228993fbebfdd11","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-13 18:30:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be4f9afe0b4f934a89b9cedf6c216196044166e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 18:39:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439e34aee89b4a4ea1009bb9b306bffb : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96fd68e19f434d0da28a0778fa3b8d38 : SUCCESS in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0364c54117b545db8c1b296f7955c8ec : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f3a7be8ac4a143621b525ab955dd140163f68bc0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-13 18:41:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee833ccb43f2d79b52ecad8fe996821fba3f36ec","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-06-14 09:24:32.000000000","message":"Patch Set 1: Code-Review+1\n\nThis patch must be merged quickly as the current constraints are incompatible in OpenStack master. It breaks at least all OPNFV Functest master gates\n\nhttps://github.com/openstack/requirements/blob/master/upper-constraints.txt#L230\nhttps://github.com/openstack/requirements/blob/master/upper-constraints.txt#L416\n\nERROR: cliff 3.2.0 has requirement cmd2!\u003d0.8.3,\u003c0.9.0,\u003e\u003d0.8.0, but you\u0027ll have cmd2 1.1.0 which is incompatible.","accounts_in_message":[],"_revision_number":1},{"id":"ad83329a8fbc2d34b1accc1967e4ad8e18ca5bba","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-06-14 09:25:37.000000000","message":"Patch Set 1:\n\nConfirmed it breaks rally_openstack and all projects leveraging on it.","accounts_in_message":[],"_revision_number":1},{"id":"f4ff16bd8efcff1e0a1ebbdc9682d7150b96641f","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-06-14 09:26:59.000000000","message":"Patch Set 1:\n\nI frankly don\u0027t understand why cmd2 was updated in constraints before this patch ?!","accounts_in_message":[],"_revision_number":1},{"id":"d5edd9792ebb808da6b36e47fe26654006dc940b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-14 10:22:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a06c31a61914f2c781ff71d9316a288a6ff00068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 10:22:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"490da3c3d0978f1bf8f302122d067b90e5f78a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 10:32:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccdae401c95b467e91c96d657614cf90 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c519a1862d945938f16f16320375a72 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13d65ceaba9a43fbb8ed85c959dee2dd : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b63ea650140c2bbb985377eca720c6008268b116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 10:32:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb754272340cf22b3561c6af0722629c01619f68","revisions":{"fb754272340cf22b3561c6af0722629c01619f68":{"kind":"REWORK","_number":1,"created":"2020-06-13 18:30:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/735440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/735440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/735440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/735440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/735440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/735440/1"}}},"commit":{"parents":[{"commit":"d85d8b4e8d1ba4b23d42a4e924c93bc8dbf294bc","subject":"Release 2.2.0 of osc-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d85d8b4e8d1ba4b23d42a4e924c93bc8dbf294bc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-13 18:26:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-13 18:26:52.000000000","tz":0},"subject":"Release 3.3.0 of cliff","message":"Release 3.3.0 of cliff\n\nThis new release includes:\n- Fix compatibility with new cmd2\n- Remove cap on cmd2.\n\nThis new release will unblock the Rally CI job taks during the\nplugin discovery of \"rally_openstack\".\n\nChange-Id: I9fd1426d315e77f2dcdc5cdeeb71e35fe7781f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb754272340cf22b3561c6af0722629c01619f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb754272340cf22b3561c6af0722629c01619f68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
