)]}'
{"id":"openstack%2Freleases~737432","triplet_id":"openstack%2Freleases~master~I9cead21be5f9e84b41b164d3ab5ce2b174d12399","project":"openstack/releases","branch":"master","topic":"puppet-openstack-victoria-m1","hashtags":[],"change_id":"I9cead21be5f9e84b41b164d3ab5ce2b174d12399","subject":"Puppet OpenStack Victoria Milestone 1","status":"MERGED","created":"2020-06-23 03:32:29.000000000","updated":"2021-01-04 10:12:24.000000000","submitted":"2020-06-24 13:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737432-1593005334857-48842415","meta_rev_id":"fe1cf5dc2a4b089533c6882ef83865a25b7cad73","_number":737432,"virtual_id_number":737432,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 13:28:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-23 11:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-24 13:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-24 13:14:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-06-23 04:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 11:51:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-24 13:14:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-24 13:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 10:11:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"733ee48a976978b1f8fe5f2007fd5f615c9b67d7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-06-23 03:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b7b797c7388bbfa7a86b74690b847c5a474a1f1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 03:33:37.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"df9aa55995f001b4572de398e87232de193aef24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 03:44:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335d854146f24f64b5be9fc57831569e : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9fd8d98c06643faad83168eb1032e05 : FAILURE in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/abd7afb165614d3b8c76d91526688480 : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6f46e5500198b32b20a6995e511cb5cee4cf9353","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-06-23 04:00:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7b5897075a91755307ea7b252a57bc86e8ce5b4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 04:01:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"0eee2b0a8bf6fe9cf10f25161d9e9cab8226ea74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 04:07:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09eb5736a60466288ce1fb3b29d0b3d : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cc1bc3ae1364ca1b91b0f7cadd9e834 : SUCCESS in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/22c7122957014e79a21ba9c1723f7c4a : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f5f8afea56c1fc60f558c84d355a38cca6e4be7f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 11:51:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"36c3e1886f4e6d343f4c4d26e98ce51bcbb640c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 13:14:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9408f72d814be15b5d061aa2721c97d7ebd36125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:14:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fc1b8e56dd6d3d8d71abc56021d59116f110745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:28:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1463230009da11412a58eee03b7cc4700051dedf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:28:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a8d7e5e3dc841ddacd929d1531117c1 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeed8d3cd13c4f5498653a5403115689 : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d91606bbef074d4ba477d7212062f7d6 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1cb68d375f409b7caab205b7240674bc31b163fc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-04 10:11:29.000000000","message":"Patch Set 2:\n\nYes, for Victoria, 17.1.0 was used at milestone 1 (17.0.0 was skipped) but for Ussuri, 16.0.0 was used at milestone 1. I\u0027m not sure we should continue skipping versions as it can be confusing to users.\n\nWas there a good reason for skipping 17.0.0 in victoria, and skipping 18.0.0 in wallaby ?","accounts_in_message":[],"_revision_number":2},{"id":"fe1cf5dc2a4b089533c6882ef83865a25b7cad73","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-04 10:12:24.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Yes, for Victoria, 17.1.0 was used at milestone 1 (17.0.0 was skipped) but for Ussuri, 16.0.0 was used at milestone 1. I\u0027m not sure we should continue skipping versions as it can be confusing to users.\n\u003e \n\u003e Was there a good reason for skipping 17.0.0 in victoria, and skipping 18.0.0 in wallaby ?\n\nOooops, wrong change, please ignore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41ecfbc148bebae14b13308ac010cda3c36f4c19","revisions":{"4518428ed0d230eda0ffdb906d1e3de6d72885e2":{"kind":"REWORK","_number":1,"created":"2020-06-23 03:32:29.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/32/737432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/737432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/737432/1"}}},"commit":{"parents":[{"commit":"f8c7c937deac1c41867670b1aebe93700ece150f","subject":"Merge \"Release glance-store for victoria-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8c7c937deac1c41867670b1aebe93700ece150f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-23 03:26:15.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-23 03:31:56.000000000","tz":480},"subject":"Puppet OpenStack Victoria Milestone 1","message":"Puppet OpenStack Victoria Milestone 1\n\nChange-Id: I9cead21be5f9e84b41b164d3ab5ce2b174d12399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4518428ed0d230eda0ffdb906d1e3de6d72885e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4518428ed0d230eda0ffdb906d1e3de6d72885e2"}]},"branch":"refs/heads/master"},"41ecfbc148bebae14b13308ac010cda3c36f4c19":{"kind":"REWORK","_number":2,"created":"2020-06-23 04:00:32.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/32/737432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/737432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/737432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/737432/2"}}},"commit":{"parents":[{"commit":"f8c7c937deac1c41867670b1aebe93700ece150f","subject":"Merge \"Release glance-store for victoria-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8c7c937deac1c41867670b1aebe93700ece150f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-23 03:26:15.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-23 03:55:39.000000000","tz":480},"subject":"Puppet OpenStack Victoria Milestone 1","message":"Puppet OpenStack Victoria Milestone 1\n\nChange-Id: I9cead21be5f9e84b41b164d3ab5ce2b174d12399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/41ecfbc148bebae14b13308ac010cda3c36f4c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/41ecfbc148bebae14b13308ac010cda3c36f4c19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
