)]}'
{"id":"openstack%2Freleases~738492","triplet_id":"openstack%2Freleases~master~I7869cf412524d7352249f02f7449b2fddf401d2c","project":"openstack/releases","branch":"master","topic":"bugfix","hashtags":[],"change_id":"I7869cf412524d7352249f02f7449b2fddf401d2c","subject":"Test bugfix branching with release-test","status":"MERGED","created":"2020-06-29 19:12:30.000000000","updated":"2020-06-30 08:27:04.000000000","submitted":"2020-06-29 22:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738492-1593468612168-1a3eceda","meta_rev_id":"e2e32cb8db8fe67e6bce7bf795f01e056b7f46ee","_number":738492,"virtual_id_number":738492,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 22:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-29 22:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-29 22:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-06-29 20:41:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 19:12:44.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-06-29 22:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14dbdaeb15c8370f894d4ee96ae1171e4200a286","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 19:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e28444e3257f6ac9e7073d68c6f7ea0483c7c821","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 19:13:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3e34811acdc6ee25e75392459cc847e49e6f1c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 19:23:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc176a362df4376bda9ffa69c25d58e : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8acf203d8fd14a248065d318abe00515 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77afe89e0dd041dca5d2e3be11e97302 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe85d36e2ff748dface89e1a2742a408 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94d4db8364bb4659894d6927cf88691e : SUCCESS in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ad672a1708894302a07d30ee24ae0752 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3317a5b3f0d841d7a78ec001f45243c2 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"75bed1f916bb659e68551d59c78d8f53989c6568","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 19:56:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fadf420dbc294b96b826daf56282ee759b0f3a68","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 19:57:52.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ef625dc5cfce61939274eb0810554d4aabcbd59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 20:06:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63135c72c504b1c901adb24496864c2 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c172e352c4634d86ae94b4ab5732208f : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c9ba58643d44219bcf4e3a560e8e5e1 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c29a152a474a668a23412e190a5ce9 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/552276b609c04ec08da4fa137721f68c : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/978ac077a7c748ec8ab46ad4824567f3 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/351eb9e6d8e94b27855d25790dac5bc5 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b0e6c0d5fe162d3686ed98074867b9471c1bef2e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 20:37:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"051942ec6db18f57c75f991e24e3554a1285783c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 20:38:28.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"478c5b025dcbd25234e40ee4f4d68581c7a6b908","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 20:40:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f36c47376a7d3b9383802d81e03b581784efc1af","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 20:41:27.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"a557d245a85647a749968941b81a2953f281b6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 20:52:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359267f953734fcf946f64bd220a2691 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22c22114712b4cb588bbdbe0c58db5fe : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c61e047870d43739943ff7664cb0ac1 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce6b9d9fffc42b6ada99d1bfa253bea : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6aeab54c71e40d1b60b9269d4fda8fe : SUCCESS in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/92ffc75222be494786dd4af8eb7a3568 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63bd4843d88045c88076823e0a8ee75b : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"fe74aa472f91153ac2e5ec0c9ae6b5a482666cef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 22:00:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLet\u0027s see what breaks.","accounts_in_message":[],"_revision_number":4},{"id":"50d5afb84517a1af4311979e8e85eb2444370665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:00:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a3d6a07199599440bd83d13f24c245cb5ea83df9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:10:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d5334b85cd4a22b0d82645be46ac06 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6e1674d2adc4c94895b5fd60feff225 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a75285acfaa4675b8edb287c0eae789 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a571757dff45d8b746f63925fa805e : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c30ea3e752b48f5b174d407551356de : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df3730b7c8b145a998e0d5eb08d08b33 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"797787f373e0a1208ab4f9e0377ddc699e059e88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:10:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0c69ba4159630c8a5bd98f947d312de0e4a6c510","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-30 08:27:04.000000000","message":"Patch Set 4:\n\nI see the branch: https://opendev.org/openstack/release-test/src/branch/bugfix/3.2 \\o/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d593479b0b103f1be47ce7ec85f423aaba7e58d2","revisions":{"a9df90e9b73e606ea3d133d49bcc6492b77a6a44":{"kind":"REWORK","_number":1,"created":"2020-06-29 19:12:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/738492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/738492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/738492/1"}}},"commit":{"parents":[{"commit":"99afb5fd303ec231fc8a4acbbb1f0ce01998db3e","subject":"Merge \"Release openstacksdk 0.48.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99afb5fd303ec231fc8a4acbbb1f0ce01998db3e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:12:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:12:08.000000000","tz":-300},"subject":"WIP: Test bugfix branches","message":"WIP: Test bugfix branches\n\nChange-Id: I7869cf412524d7352249f02f7449b2fddf401d2c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9df90e9b73e606ea3d133d49bcc6492b77a6a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9df90e9b73e606ea3d133d49bcc6492b77a6a44"}]},"branch":"refs/heads/master"},"82bbb31ef41684bd2b1d6b210f7a16281a38dd24":{"kind":"REWORK","_number":2,"created":"2020-06-29 19:56:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/738492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/738492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/738492/2"}}},"commit":{"parents":[{"commit":"99afb5fd303ec231fc8a4acbbb1f0ce01998db3e","subject":"Merge \"Release openstacksdk 0.48.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99afb5fd303ec231fc8a4acbbb1f0ce01998db3e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:12:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:56:27.000000000","tz":-300},"subject":"WIP: Test bugfix branches","message":"WIP: Test bugfix branches\n\nChange-Id: I7869cf412524d7352249f02f7449b2fddf401d2c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/82bbb31ef41684bd2b1d6b210f7a16281a38dd24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/82bbb31ef41684bd2b1d6b210f7a16281a38dd24"}]},"branch":"refs/heads/master"},"2fcb75005689b8afec989f110b249a1f189f4fe4":{"kind":"REWORK","_number":3,"created":"2020-06-29 20:37:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/738492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/738492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/738492/3"}}},"commit":{"parents":[{"commit":"99afb5fd303ec231fc8a4acbbb1f0ce01998db3e","subject":"Merge \"Release openstacksdk 0.48.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99afb5fd303ec231fc8a4acbbb1f0ce01998db3e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:12:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 20:36:27.000000000","tz":-300},"subject":"Test bugfix branching with release-test","message":"Test bugfix branching with release-test\n\nThis does a new release with release-test to see if everything is\ncovered to allow branches to have a bugfix/* prefix.\n\nChange-Id: I7869cf412524d7352249f02f7449b2fddf401d2c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fcb75005689b8afec989f110b249a1f189f4fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fcb75005689b8afec989f110b249a1f189f4fe4"}]},"branch":"refs/heads/master"},"d593479b0b103f1be47ce7ec85f423aaba7e58d2":{"kind":"REWORK","_number":4,"created":"2020-06-29 20:40:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/738492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/738492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/738492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/738492/4"}}},"commit":{"parents":[{"commit":"99afb5fd303ec231fc8a4acbbb1f0ce01998db3e","subject":"Merge \"Release openstacksdk 0.48.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99afb5fd303ec231fc8a4acbbb1f0ce01998db3e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 19:12:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 20:40:12.000000000","tz":-300},"subject":"Test bugfix branching with release-test","message":"Test bugfix branching with release-test\n\nThis does a new release with release-test to see if everything is\ncovered to allow branches to have a bugfix/* prefix.\n\nChange-Id: I7869cf412524d7352249f02f7449b2fddf401d2c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d593479b0b103f1be47ce7ec85f423aaba7e58d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d593479b0b103f1be47ce7ec85f423aaba7e58d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
