)]}'
{"id":"openstack%2Freleases~739686","triplet_id":"openstack%2Freleases~master~I36461375db7a55aeb9cd87acf87870cb6db8f543","project":"openstack/releases","branch":"master","topic":"castellan-ussuri-07-07","hashtags":[],"change_id":"I36461375db7a55aeb9cd87acf87870cb6db8f543","subject":"Release castellan 3.0.2 for ussuri","status":"MERGED","created":"2020-07-07 09:19:48.000000000","updated":"2020-07-09 09:00:11.000000000","submitted":"2020-07-09 09:00:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739686-1594285211119-f3086ce0","meta_rev_id":"bee5dead6b926e424ffc486644dbe38dd9b6fe25","_number":739686,"virtual_id_number":739686,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-07-07 09:53:03.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-09 09:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-07 20:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-09 08:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-09 08:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-07 20:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 20:22:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-09 08:49:00.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-09 09:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c419bf804eb35d084c213302d0efd201bb7bcb7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 09:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c7de7f47e41328e816cf43a6acbb899049f9e70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:25:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"08d43f4e1d6ef09f11cb0edc3333ccc24036c2f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:31:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d628ed4d8646629ea86ca963ca0b5c : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba8c56d1325a4caca933cd11f09a3051 : FAILURE in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/37d25e3b97034a3287d21a5d63025342 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"42aee95de44871eecdf131d82b5029021f4e2f60","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 09:46:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63d696456d2808a062ad4374fb0088c407e58a9e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:48:06.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"db0770a1b40fee9b0b35a52fab1c49abb551371a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 09:51:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5d856fcf99e2323dd17775094537795df1f5ab5e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:52:48.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"402ceda7c005798b01ca5bbb0778244fafbab7e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 09:53:03.000000000","message":"Patch Set 3:\n\nCastellan 3.1.0 is already in use on victoria so move to 3.0.2","accounts_in_message":[],"_revision_number":3},{"id":"d99081de420c12a5e2239b72fe2b3a1b3a862d30","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 09:53:12.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":3},{"id":"1afe324e116f24a11c59f008e805461aadf668af","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:54:27.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dd36dbe50e0cd5972fb4936718fd6a0067be0500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:07:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fdf4fc2fe24e27a206d29aecb4fa20 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85519fe261924180bbd3a64d37be3d19 : SUCCESS in 6m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/24e34331581c4df3a4fdd46dcfc32ae6 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"fa88728bdf394da01d698114069f5ee66b5bbc80","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 12:15:00.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":3},{"id":"8a35b879982785c7cde1d31313f0b9760fab5b65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-07 20:22:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"faf7aba09f2c39d1d358724fe0bfc09b253e219a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:24:22.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a06a3f940c7ea14c889173266ebadcc6dc14061f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-09 08:49:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c1d515baa52a3579391f0ade5d9a3ab68a308915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:49:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea788b7a1dea449100867cc62161fe7fe5573604","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:00:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eab0c27436644fa8d9e78ab739c0f59 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3086e4a55bf45a2919ac02bc657b7a3 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/395d1c5d64cf4698be962ce654d3c5aa : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"5d0af580a348aa479e09162a12bff644b4adfd3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:00:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da98cfefd033876ec86d4d2335f16d2d09821176","revisions":{"05de52bb429ae59c089069b847f8a5b75c0d13e4":{"kind":"REWORK","_number":1,"created":"2020-07-07 09:19:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/739686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/739686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/739686/1"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:15:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:15:05.000000000","tz":120},"subject":"Release castellan 3.1.0 for ussuri","message":"Release castellan 3.1.0 for ussuri\n\nA new CLI option was added to verify the SSL path and solve an issue [1]\nthis is why I proposed a minor version on stable branch.\n\nIf this option is not given by operator then the behavior stay\nconsistent with previous versions.\n\n[1] https://review.opendev.org/#/c/738399/\n\nChange-Id: I36461375db7a55aeb9cd87acf87870cb6db8f543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05de52bb429ae59c089069b847f8a5b75c0d13e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05de52bb429ae59c089069b847f8a5b75c0d13e4"}]},"branch":"refs/heads/master"},"930bc12fc94c510f5da7335d9b4f9362d0b36d0b":{"kind":"REWORK","_number":2,"created":"2020-07-07 09:46:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/739686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/739686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/739686/2"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:15:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:46:13.000000000","tz":120},"subject":"Release castellan 3.1.0 for ussuri","message":"Release castellan 3.1.0 for ussuri\n\nA new CLI option was added to verify the SSL path and solve an issue [1]\nthis is why I proposed a minor version on stable branch.\n\nIf this option is not given by operator then the behavior stay\nconsistent with previous versions.\n\n[1] https://review.opendev.org/#/c/738399/\n\nChange-Id: I36461375db7a55aeb9cd87acf87870cb6db8f543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/930bc12fc94c510f5da7335d9b4f9362d0b36d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/930bc12fc94c510f5da7335d9b4f9362d0b36d0b"}]},"branch":"refs/heads/master"},"da98cfefd033876ec86d4d2335f16d2d09821176":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-07 09:51:18.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/739686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/739686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/739686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/739686/3"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:15:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 09:46:55.000000000","tz":120},"subject":"Release castellan 3.0.2 for ussuri","message":"Release castellan 3.0.2 for ussuri\n\nA new CLI option was added to verify the SSL path and solve an issue\n[1].\n\nUnfortunattely we can\u0027t propose a minor release because 3.1.0 is already\nin use on victoria [2].\n\nAlso backport new feature and options on stable is something that we\nwant to avoid but here if this option is not given by operator then the\nbehavior stay consistent with previous versions, so I don\u0027t think it can\nhurt.\n\n[1] https://review.opendev.org/#/c/738399/\n[2] https://opendev.org/openstack/releases/src/branch/master/deliverables/victoria/castellan.yaml#L9\n\nChange-Id: I36461375db7a55aeb9cd87acf87870cb6db8f543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da98cfefd033876ec86d4d2335f16d2d09821176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da98cfefd033876ec86d4d2335f16d2d09821176"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
