)]}'
{"id":"openstack%2Freleases~739714","triplet_id":"openstack%2Freleases~master~I95e5eefda3fb950859b9c72528977ad2aaf6b0f5","project":"openstack/releases","branch":"master","topic":"bifrost-victoria-820","hashtags":[],"change_id":"I95e5eefda3fb950859b9c72528977ad2aaf6b0f5","subject":"Release bifrost 8.2.0 for victoria","status":"MERGED","created":"2020-07-07 11:20:55.000000000","updated":"2020-07-08 13:40:19.000000000","submitted":"2020-07-08 13:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739714-1594215619745-86bf99d6","meta_rev_id":"ab81805079dcf69daf8911cc9136a24048aff478","_number":739714,"virtual_id_number":739714,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 13:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-08 13:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-08 12:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-08 13:20:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-08 12:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 09:19:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-08 12:51:06.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-07-08 13:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-08 13:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fcee2acb49e9e8a40a430f22df5c850ccc7303a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-07 11:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"569dfe2123f612e2d13d9e8461eaffb1af4c8a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 11:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18d365b07d14375a5996d0b4524f177 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d82bfeea6d147668d59b0232df919d6 : SUCCESS in 7m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7cf70835c28f41db8cd34c0f92eb9a3b : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"295f2a276ad9d9e2c2f8fd79dde1121b32b1aef8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-07 20:32:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff6e1f12d08bf4fb084548c867065d296fb6c5d5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-08 08:21:00.000000000","message":"Patch Set 1: Code-Review-1\n\nPer our new release model, this needs a bugfix/8.2 branch, I guess?","accounts_in_message":[],"_revision_number":1},{"id":"b881c98df580ad34956af0d5ffda22848ed4551c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-08 09:06:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"498fbe390ff0f4ffce71e0cff76e4cea51ca9b20","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-08 09:07:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"76f2276c1d2fb60260b5dcfcfcb8f2e118abb588","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:19:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eee32520d54ed6898c5a166243dfe235900e6320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:25:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f416352c167f474ca4633edf6a384fa2 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b194ab83ae94051b81a3c74c25d4f9b : SUCCESS in 6m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/38a496437d1043f79ffd4dd47f1e61e7 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"06d7c3bafdd8ff04ae86542ae1fed17c48072f28","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-08 09:25:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cbb9c47db39d2e62193b4943b750c901da6d151","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-08 12:29:06.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ecf49d04c3572952abef47424a0b832f60c81e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 12:50:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554a4aea70174653bd508638ae643598 : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c96b4dbeff3148aaaa63e99c651d9ee9 : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/299f18e550e040cbb12b5f914f13c0bc : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":4},{"id":"54b88812f9a3d0e2fb7fd7ea94f528d888dc4c88","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-08 12:51:06.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks correct now, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"5ec4d619348f14985a935728fad4078e0163d0fc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 12:52:51.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"a195d5eafa54975b607661cb0813e884a4118a98","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 13:20:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"360ef24efd56ec2be4f4b67ef8c733bb4122639d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:20:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6412f29f45028fe7048761162e31003b4a6a6c5e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 13:20:41.000000000","message":"Patch Set 4:\n\nWe still need to figure out how to do releases from these bugfix branches yet, but at least the branch can be created.","accounts_in_message":[],"_revision_number":4},{"id":"9b6dc7f696b17eaea324a8d4a74e9a56ab5ebe3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:40:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"30ca6154c5c60dc6f9e452aea2bf1d5aadac53e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:40:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451138766fdc4dafa0b110e72182570e : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f077f9de9e848bf8b58597673ca5845 : SUCCESS in 10m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7d9d10dce594cbb8d128cb07f973e51 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2028b7434f1bd36e0d2c22d714f5df9dfbc5621b","revisions":{"d2f3b1aa848395efab6e247b8a7865f3b58e186d":{"kind":"REWORK","_number":1,"created":"2020-07-07 11:20:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/14/739714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/739714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/739714/1"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-07 11:20:29.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-07 11:20:29.000000000","tz":120},"subject":"Release bifrost 8.2.0 for victoria","message":"Release bifrost 8.2.0 for victoria\n\nChange-Id: I95e5eefda3fb950859b9c72528977ad2aaf6b0f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2f3b1aa848395efab6e247b8a7865f3b58e186d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2f3b1aa848395efab6e247b8a7865f3b58e186d"}]},"branch":"refs/heads/master"},"a6fb1311bf91a35168a367cbac858eea92873107":{"kind":"REWORK","_number":2,"created":"2020-07-08 09:06:49.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/14/739714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/739714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/739714/2"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-07 11:20:29.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-08 09:06:13.000000000","tz":120},"subject":"Release bifrost 8.2.0 for victoria","message":"Release bifrost 8.2.0 for victoria\n\nAlso add stable-branch-type: std-with-versions to reflect change\nin the release model.\n\nChange-Id: I95e5eefda3fb950859b9c72528977ad2aaf6b0f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6fb1311bf91a35168a367cbac858eea92873107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6fb1311bf91a35168a367cbac858eea92873107"}]},"branch":"refs/heads/master"},"301e48c9c3abb0418802c96dfdb35b5bfdf4c2c4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-08 09:07:36.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/14/739714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/739714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/739714/3"}}},"commit":{"parents":[{"commit":"339af5321ac03580bc1288fab7d010693b65ed00","subject":"Fix tox command in tool help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/339af5321ac03580bc1288fab7d010693b65ed00"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-07 11:20:29.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-08 09:07:27.000000000","tz":120},"subject":"Release bifrost 8.2.0 for victoria","message":"Release bifrost 8.2.0 for victoria\n\nAlso add stable-branch-type: std-with-versions to reflect change\nin the release model.\n\nChange-Id: I95e5eefda3fb950859b9c72528977ad2aaf6b0f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301e48c9c3abb0418802c96dfdb35b5bfdf4c2c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301e48c9c3abb0418802c96dfdb35b5bfdf4c2c4"}]},"branch":"refs/heads/master"},"2028b7434f1bd36e0d2c22d714f5df9dfbc5621b":{"kind":"REWORK","_number":4,"created":"2020-07-08 12:29:06.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/14/739714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/739714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/739714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/739714/4"}}},"commit":{"parents":[{"commit":"339af5321ac03580bc1288fab7d010693b65ed00","subject":"Fix tox command in tool help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/339af5321ac03580bc1288fab7d010693b65ed00"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-07 11:20:29.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-08 12:29:00.000000000","tz":0},"subject":"Release bifrost 8.2.0 for victoria","message":"Release bifrost 8.2.0 for victoria\n\nAlso add stable-branch-type: std-with-versions to reflect change\nin the release model.\n\nChange-Id: I95e5eefda3fb950859b9c72528977ad2aaf6b0f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2028b7434f1bd36e0d2c22d714f5df9dfbc5621b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2028b7434f1bd36e0d2c22d714f5df9dfbc5621b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
