)]}'
{"id":"openstack%2Freleases~739818","triplet_id":"openstack%2Freleases~master~Iaf86ecb1589c40102acb621b23ea12d71ed453bb","project":"openstack/releases","branch":"master","topic":"rewrite_list_unreleased_changes","hashtags":[],"change_id":"Iaf86ecb1589c40102acb621b23ea12d71ed453bb","subject":"Allow list_unreleased_changes to format results at json \u0026 yaml formats","status":"MERGED","created":"2020-07-07 16:58:37.000000000","updated":"2020-07-10 16:05:02.000000000","submitted":"2020-07-10 16:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739818-1594397102945-a252ec55","meta_rev_id":"893a0559b7f1403a92296949e194b385dd1c0d2b","_number":739818,"virtual_id_number":739818,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-07-09 09:19:44.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 16:05:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 15:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-09 08:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-10 15:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 08:59:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-10 15:47:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 16:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25f91ac6a15d3a0fd3c282e0daa0dc23f559ff09","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 16:58:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3b184ee38b3746168a33dace8397d9b314cb1ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 17:04:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9842b84b3cee1d9db33eb5a6ebe6bfede82963a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 17:06:41.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e3f9289bc11b0ebc02de91f4434cc61e69960a99","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-07 17:08:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8cf8dc224e81147a746fcd05562447518cec48e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:22:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1dae6442f74250bd5ca612e3a5d6a2 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2a1c5ad41e4293baa9a013f2a4bc29 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d35fce494446fc9ca366250873b7d7 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da063517167b48538df06176d8966fa7 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b6335933ca74c07b04d4aee8fcb3a69 : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/61606470de6f4bee9e8c63daeaafe282 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f6a8a503d07483bb856f55786b2aa27 : SUCCESS in 5m 38s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/417d583363a24511903ae6cf18d0dab8 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":4},{"id":"021fdc033a082c6c4b663b961a6ef1ed33934290","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-09 08:59:36.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, we can easily iterate on fixes if needed.\n\nSeparating making it a Python script and adding new features would have made the review slightly easier :)","accounts_in_message":[],"_revision_number":4},{"id":"c768f65345126bde687b3e31252fe2e0e3b95c31","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 09:19:44.000000000","message":"Patch Set 4:\n\n\u003e LGTM, we can easily iterate on fixes if needed.\n \u003e \n \u003e Separating making it a Python script and adding new features would\n \u003e have made the review slightly easier :)\n\nYou\u0027re right sorry, this how I was thinking work in a first but finally I done it in a all in one patch.","accounts_in_message":[],"_revision_number":4},{"id":"409eeaad784fc5ba5adee74c40e28e3487b6207b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:47:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5c2abfd539cd45ad14e10b67fd709648c0450e45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:49:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0679746d787b4dea94c3bc9a5a94bc0a34e2705c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:05:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"152c230d86509cb0420e1297d3c18aaf0c85b5e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:05:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f57f356b714fb4886d5581881d463f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c372934cc91429f91e72775400c7c31 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6edb17401d95472e8b83a7cedb187ba0 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a05db89de1f4d0f800cfdcab458ecbb : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47b680c5c8704e4f9af9939201dad755 : SUCCESS in 11m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb9df2f08cda4814b7ec708353892f2e : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4912f7d5d0a60b09ea54a5e6835c00847455f2dd","revisions":{"e040679956b180f343b77c9469d38316944d6195":{"kind":"REWORK","_number":1,"created":"2020-07-07 16:58:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/739818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/739818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/739818/1"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 15:36:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 16:58:24.000000000","tz":120},"subject":"Rewrite list unreleased changes in python","message":"Rewrite list unreleased changes in python\n\nBy default it will generate the same output that generated previously\nby the shell script version of this tools.\n\nShell script is still there but it will call the python command in a\nvenv instead of directly implement features.\n\nFeatures:\n- allow user to retrieve results in json format\n- allow user to retrieve results in yaml format\n- allow user to ignore project not yet released\n\nChange-Id: Iaf86ecb1589c40102acb621b23ea12d71ed453bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e040679956b180f343b77c9469d38316944d6195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e040679956b180f343b77c9469d38316944d6195"}]},"branch":"refs/heads/master"},"d684a6758d42f5ec8eb1913b2685725c2d6773b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-07 17:04:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/739818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/739818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/739818/2"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 15:36:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 17:01:45.000000000","tz":120},"subject":"Rewrite list unreleased changes in python","message":"Rewrite list unreleased changes in python\n\nBy default it will behave as the previous version of this command (the\nshell script).\n\nFew new feature have been added in these changes to allow us to more\neasily handle outputs in scripts.\n\nFeatures added:\n- allow user to retrieve results in json format\n- allow user to retrieve results in yaml format\n- allow user to ignore project not yet released\n\nThe shell script entry-point (tools) is still there but it will call\nthe python command in a venv instead of directly implement features.\n\nChange-Id: Iaf86ecb1589c40102acb621b23ea12d71ed453bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d684a6758d42f5ec8eb1913b2685725c2d6773b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d684a6758d42f5ec8eb1913b2685725c2d6773b2"}]},"branch":"refs/heads/master"},"7444ab307600c90ca9531f6909af1b8ad93f4cff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-07 17:06:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/739818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/739818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/739818/3"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 15:36:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 17:04:34.000000000","tz":120},"subject":"Allow list_unreleased_changes to format results at json \u0026 yaml formats","message":"Allow list_unreleased_changes to format results at json \u0026 yaml formats\n\nRewrite list_unreleased_changes as python format and add new features.\n\nBy default it will behave as the previous version of this command (the\nshell script).\n\nFew new feature have been added in these changes to allow us to more\neasily handle outputs in scripts.\n\nFeatures added:\n- allow user to retrieve results in json format\n- allow user to retrieve results in yaml format\n- allow user to ignore project not yet released\n\nThe shell script entry-point (tools) is still there but it will call\nthe python command in a venv instead of directly implement features.\n\nChange-Id: Iaf86ecb1589c40102acb621b23ea12d71ed453bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7444ab307600c90ca9531f6909af1b8ad93f4cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7444ab307600c90ca9531f6909af1b8ad93f4cff"}]},"branch":"refs/heads/master"},"4912f7d5d0a60b09ea54a5e6835c00847455f2dd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-07-07 17:08:35.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/739818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/739818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/739818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/739818/4"}}},"commit":{"parents":[{"commit":"0ee0425e891fe9e890ff12af1cfd60670ccea3df","subject":"Merge \"Release keystoneauth 3.17.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ee0425e891fe9e890ff12af1cfd60670ccea3df"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 15:36:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-07 17:07:25.000000000","tz":120},"subject":"Allow list_unreleased_changes to format results at json \u0026 yaml formats","message":"Allow list_unreleased_changes to format results at json \u0026 yaml formats\n\nRewrite list_unreleased_changes as python format and add new features.\n\nBy default it will behave as the previous version of this command (the\nshell script).\n\nFew new feature have been added in these changes to allow us to more\neasily handle outputs in scripts.\n\nFeatures added:\n- allow user to retrieve results in json format\n- allow user to retrieve results in yaml format\n- allow user to ignore project not yet released\n\nThe shell script entry-point (tools) is still there but it will call\nthe python command in a venv instead of directly implement features.\n\nAlso the python version allow us to more surround this tools with unit tests.\n\nChange-Id: Iaf86ecb1589c40102acb621b23ea12d71ed453bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4912f7d5d0a60b09ea54a5e6835c00847455f2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4912f7d5d0a60b09ea54a5e6835c00847455f2dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
