)]}'
{"id":"openstack%2Freleases~740512","triplet_id":"openstack%2Freleases~master~I901c31fae1ce870819682b1eaf34d6768e61d960","project":"openstack/releases","branch":"master","topic":"osc-performance","hashtags":[],"change_id":"I901c31fae1ce870819682b1eaf34d6768e61d960","subject":"stevedore 3.0.0","status":"MERGED","created":"2020-07-10 17:55:06.000000000","updated":"2020-07-10 19:55:08.000000000","submitted":"2020-07-10 19:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740512-1594410908669-1939c49e","meta_rev_id":"3db43432ad69d8eba6d8833ffc6dc19f49302bb0","_number":740512,"virtual_id_number":740512,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 19:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-07-10 19:27:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-07-10 19:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-07-10 19:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-10 19:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 19:27:34.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-10 19:42:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 19:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6aaaa54f006dc66c2f3cd2514c0f417e5b27df27","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-10 17:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"092d2c47c5dc97b0524c952e84284c40bf0b096b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:05:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed977878c2c472b80141364a44da1aa : SUCCESS in 9m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cd3dcf26c214f53b161398562039405 : SUCCESS in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/95fbc98c84fc49c29f84cf169e004ca6 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e1343ce8e86d92439ca609f2403c0bb15b372b4c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 18:10:50.000000000","message":"Patch Set 1: Code-Review+2\n\nAck from Ben in IRC, but will wait a bit to get that recorded here.","accounts_in_message":[],"_revision_number":1},{"id":"c359b72a10df1e9413e78c25a151f9a74badd95a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-10 18:47:02.000000000","message":"Patch Set 1:\n\nDo we need to bump the hash so the reno gets associated with the right release?","accounts_in_message":[],"_revision_number":1},{"id":"fde6075997541336c83469948f886e8e9cccdee2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 18:47:52.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"a7bd4f179da987ea8d6a4e8f5da2d831f818b158","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 19:04:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4132ba9c45aeaafd82e475b7704fab4e3f9164d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:14:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb53efc42a44fdea3fb2a28ac4f7317 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61423cf7a8494a78bdce311196495c38 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7ff616d14924b16a50128f628ef4fa3 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b6e7a171c41dde935d6965b3e2d3288f48f7dd43","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-10 19:27:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6c01c8e92d7fa7346268c9a8792151b6acb46e30","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:28:43.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a63e805daf134755f53bc6b63a3df22d6dcd788f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 19:42:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68dcffd5ebea535ea88c5cc84adc6bd687d2df23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:42:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9319ec02d11f8f20b38bc91c445066d523fc59c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:55:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55231e46ec484875b76a3cfaf6c8f241 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c30a46cab94bb6aad80280228ef017 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1d1b27b91d14a2bad18a6b3dd85845d : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"12c3c6bfe57cd3cac79418ac3f032a7e5ffb8af5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f54b1a1a63d9de85484364de8dee69e47f1dec66","revisions":{"b241258f1b30e237b36903460e9b609bdc03af77":{"kind":"REWORK","_number":1,"created":"2020-07-10 17:55:06.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/12/740512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/740512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/740512/1"}}},"commit":{"parents":[{"commit":"baa0151db1c8039527554c6b7b872f21bf81c005","subject":"Merge \"Move Validations repos from ussuri to independent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/baa0151db1c8039527554c6b7b872f21bf81c005"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-10 17:53:58.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-10 17:53:58.000000000","tz":-240},"subject":"stevedore 3.0.0","message":"stevedore 3.0.0\n\nRaise the major version because the new release will return a different\ntype of EntryPoint object which is owned by another library and has a\ndifferent API from the type that used to be used.\n\nChange-Id: I901c31fae1ce870819682b1eaf34d6768e61d960\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b241258f1b30e237b36903460e9b609bdc03af77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b241258f1b30e237b36903460e9b609bdc03af77"}]},"branch":"refs/heads/master"},"f54b1a1a63d9de85484364de8dee69e47f1dec66":{"kind":"REWORK","_number":2,"created":"2020-07-10 19:04:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/12/740512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/740512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/740512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/740512/2"}}},"commit":{"parents":[{"commit":"baa0151db1c8039527554c6b7b872f21bf81c005","subject":"Merge \"Move Validations repos from ussuri to independent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/baa0151db1c8039527554c6b7b872f21bf81c005"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-10 19:04:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-10 19:04:15.000000000","tz":-300},"subject":"stevedore 3.0.0","message":"stevedore 3.0.0\n\nRaise the major version because the new release will return a different\ntype of EntryPoint object which is owned by another library and has a\ndifferent API from the type that used to be used.\n\nChange-Id: I901c31fae1ce870819682b1eaf34d6768e61d960\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f54b1a1a63d9de85484364de8dee69e47f1dec66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f54b1a1a63d9de85484364de8dee69e47f1dec66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
