)]}'
{"id":"openstack%2Freleases~741212","triplet_id":"openstack%2Freleases~master~I18ab5d7afa350746ba2310a219755f07ef957b06","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I18ab5d7afa350746ba2310a219755f07ef957b06","subject":"Create validations-libs and validations-common min version","status":"MERGED","created":"2020-07-15 13:59:38.000000000","updated":"2020-07-22 14:02:33.000000000","submitted":"2020-07-22 14:02:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741212-1595426553370-798a8ba2","meta_rev_id":"c942a6ae2d938fab5a1db878a892908baf696d59","_number":741212,"virtual_id_number":741212,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 14:02:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-16 12:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2020-07-16 11:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-07-22 13:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-07-17 09:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-16 10:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-22 13:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-16 12:40:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-22 13:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 14:15:28.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-07-16 10:37:24.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-07-16 11:03:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-16 12:40:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-17 09:55:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-22 13:27:48.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-07-22 13:30:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 14:02:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87939957088cebc799a8f215b55816b92d27ad88","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-07-15 13:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"618c8a011234416e40106ce048289000cb41de6e","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-07-15 14:08:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"993946a198b3cc4a3959a63029fde7607b35a998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 14:22:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ebae96740f4204b73850627a6bd348 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e39b8f7f907418bb0627ffdb3556b23 : SUCCESS in 9m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46488efa827e4df6af64ab49ef164608 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f3c30b83a6c4b5d80aaa38a8099f337e038e1b3c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 15:15:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a376183f9a341a61c2470dbcbda338010f25e19","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-07-15 16:08:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76acafad33e6e918710c5cac22041661746c5843","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 16:20:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a2f1c96ee6f53615c02972298cceee1b5cd4142","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-07-16 07:41:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e910f99518067af756ca089b08f08aff3e20f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:11:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30621bb333404e6cb1ca7598967bd2da : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a68870b153a24aa4b1c52a1a86536cd9 : SUCCESS in 6m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/52976d51fffe47a7bb917a5a98b93559 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"868f3e0730e40ae9c2d5a293b962fc708d48e13a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-07-16 09:42:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1370c9a16c0e6eaf1cd50bbf66da0fa74ba93113","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-07-16 09:47:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"506a7fcf35653de56d7643419e7db19a99ab89a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 10:01:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4c166f88fd43d09280b96e1b2051dc : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc501b6894dd46c6a5142c33b9ee7801 : SUCCESS in 8m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a160e0b0dcf48e8b6a9e84cc2b8b334 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":4},{"id":"426bfc03bd076e7183fb7a1068ff4cbcf5848f88","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-07-16 10:37:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0d8d0f3874269196225dd08c5811d8138cb88f4c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-16 11:03:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"61c518f50e9dc33a49776372f680a5cfae70fe6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-16 12:40:38.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks good, just waiting for PTL ack.","accounts_in_message":[],"_revision_number":4},{"id":"8f388c369e9598c833d0adb5a42a2d7b2ed2ae91","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-17 09:55:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"14da2bc5e017b88597343bf372f39bcba8991fe5","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 13:27:48.000000000","message":"Patch Set 4: Code-Review+1\n\nthanks!","accounts_in_message":[],"_revision_number":4},{"id":"0b034cc74fb48a9143fc1f4fa9c4f51a056a07b4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:28:59.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"2c6d160404d5322a2562fd175eee561d03634607","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 13:30:27.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f0782a6e4badd450befd45519daafdb9ad2826ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:30:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"67ca34c146e0a80424cb6e7520132328269fcb9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:02:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd06ba399f644a9d8f5c23dc458412c7 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f7a0270c18414a8c49ed1b62b40947 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7b87f08374e4fe6b425d1b409cb6e32 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":4},{"id":"58e29b7a67e40d5a63ddcd78878570cb67c96d16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:02:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e49a68ac8ed84d2935efb51c7974ae13057f4f04","revisions":{"b797b14dcd4160a58c3c035d921413283174c440":{"kind":"REWORK","_number":1,"created":"2020-07-15 13:59:38.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/12/741212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/741212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/741212/1"}}},"commit":{"parents":[{"commit":"d122285d0324b926aaefc11399858fc78936d5ae","subject":"Merge \"Release kolla stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d122285d0324b926aaefc11399858fc78936d5ae"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 13:58:42.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 13:58:42.000000000","tz":120},"subject":"Create validations-libs and validations-common min version","message":"Create validations-libs and validations-common min version\n\nAdd validations-libs 1.0.2 min version\nAdd validations-common 1.0.1 min version\n\nChange-Id: I18ab5d7afa350746ba2310a219755f07ef957b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b797b14dcd4160a58c3c035d921413283174c440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b797b14dcd4160a58c3c035d921413283174c440"}]},"branch":"refs/heads/master"},"05fb2efda3354536f8427f1e34880e00d17aaebf":{"kind":"REWORK","_number":2,"created":"2020-07-15 14:08:26.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/12/741212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/741212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/741212/2"}}},"commit":{"parents":[{"commit":"d122285d0324b926aaefc11399858fc78936d5ae","subject":"Merge \"Release kolla stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d122285d0324b926aaefc11399858fc78936d5ae"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 13:58:42.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 14:08:14.000000000","tz":120},"subject":"Create validations-libs and validations-common min version","message":"Create validations-libs and validations-common min version\n\nAdd validations-libs 1.0.2 min version\nAdd validations-common 1.0.1 min version\n\nChange-Id: I18ab5d7afa350746ba2310a219755f07ef957b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05fb2efda3354536f8427f1e34880e00d17aaebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05fb2efda3354536f8427f1e34880e00d17aaebf"}]},"branch":"refs/heads/master"},"ef4b99aa1d5abd80c847fe5f9e357ca98220637f":{"kind":"REWORK","_number":3,"created":"2020-07-16 07:41:09.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/12/741212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/741212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/741212/3"}}},"commit":{"parents":[{"commit":"d122285d0324b926aaefc11399858fc78936d5ae","subject":"Merge \"Release kolla stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d122285d0324b926aaefc11399858fc78936d5ae"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 13:58:42.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-16 07:40:55.000000000","tz":120},"subject":"Create validations-libs and validations-common min version","message":"Create validations-libs and validations-common min version\n\nAdd validations-libs 1.0.2 min version\nAdd validations-common 1.0.1 min version\n\nChange-Id: I18ab5d7afa350746ba2310a219755f07ef957b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef4b99aa1d5abd80c847fe5f9e357ca98220637f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef4b99aa1d5abd80c847fe5f9e357ca98220637f"}]},"branch":"refs/heads/master"},"e49a68ac8ed84d2935efb51c7974ae13057f4f04":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-07-16 09:47:38.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/12/741212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/741212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/741212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/741212/4"}}},"commit":{"parents":[{"commit":"d122285d0324b926aaefc11399858fc78936d5ae","subject":"Merge \"Release kolla stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d122285d0324b926aaefc11399858fc78936d5ae"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-07-15 13:58:42.000000000","tz":120},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-07-16 09:47:34.000000000","tz":0},"subject":"Create validations-libs and validations-common min version","message":"Create validations-libs and validations-common min version\n\nAdd validations-libs 1.0.2 min version\nAdd validations-common 1.1.0 min version\n\nChange-Id: I18ab5d7afa350746ba2310a219755f07ef957b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e49a68ac8ed84d2935efb51c7974ae13057f4f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e49a68ac8ed84d2935efb51c7974ae13057f4f04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
