)]}'
{"id":"openstack%2Freleases~741733","triplet_id":"openstack%2Freleases~master~Idc785c297104f7594f94a6b2b36bef8bc403afba","project":"openstack/releases","branch":"master","topic":"stein-stable","hashtags":[],"change_id":"Idc785c297104f7594f94a6b2b36bef8bc403afba","subject":"Release cloudkitty for stable/stein","status":"ABANDONED","created":"2020-07-17 22:05:53.000000000","updated":"2020-07-27 19:40:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32bdb1e25ccfceff035261fd97c17334b8d09309","_number":741733,"virtual_id_number":741733,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-18 00:17:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 00:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abf51e3e8a9e2c17ba495c453f761ce6d0bc0fe6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 22:05:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2321319fa95eb492c7596073748169bba3635784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 00:17:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3f6b3505a5402ea9bfd6d796c3b121 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa18d5a160ef4585ba117aa2df8fa9cc : SUCCESS in 6m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/477e1de694a34f0da7a1cbeadc6ad116 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"88ea2cb00ea01793157f150e5a51180caaa960f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-27 19:40:11.000000000","message":"Abandoned\n\nNo response, and release not required.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a604d964b7dd7471eb902ff71c68ed5e4a0c8cba","revisions":{"a604d964b7dd7471eb902ff71c68ed5e4a0c8cba":{"kind":"REWORK","_number":1,"created":"2020-07-17 22:05:53.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/33/741733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/741733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/741733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/741733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/741733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/741733/1"}}},"commit":{"parents":[{"commit":"a0128974b0c77a65f3bc26d87ab42bd46d22d99c","subject":"Release oslo\u0027s projects for victoria - 2020-07-17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0128974b0c77a65f3bc26d87ab42bd46d22d99c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:05:52.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:05:52.000000000","tz":-300},"subject":"Release cloudkitty for stable/stein","message":"Release cloudkitty for stable/stein\n\nThis release picks up new commits to cloudkitty since\nthe last release from stable/stein.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nThis is not a release team activity, it is just to help out with\nstable releases. If there is no response to the patch, we will\nassume a release is not needed and abandon the patch after one\nweek.\n\nChange-Id: Idc785c297104f7594f94a6b2b36bef8bc403afba\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a604d964b7dd7471eb902ff71c68ed5e4a0c8cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a604d964b7dd7471eb902ff71c68ed5e4a0c8cba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
