)]}'
{"id":"openstack%2Freleases~741760","triplet_id":"openstack%2Freleases~master~I3d6e9cc1aaafa640bdd884278d0b7e41d187074b","project":"openstack/releases","branch":"master","topic":"stein-stable","hashtags":[],"change_id":"I3d6e9cc1aaafa640bdd884278d0b7e41d187074b","subject":"nova: Release 19.3.0","status":"MERGED","created":"2020-07-17 22:25:09.000000000","updated":"2020-07-24 14:50:31.000000000","submitted":"2020-07-24 14:50:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741760-1595602231743-0ec4c45c","meta_rev_id":"461b2568e0c541d12a730fe39adc4d72ae972d4d","_number":741760,"virtual_id_number":741760,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-24 08:10:44.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 14:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 14:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-07-24 10:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2020-07-24 09:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-24 08:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-24 14:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-24 08:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 23:20:35.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-07-24 08:10:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-24 08:13:03.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-07-24 09:01:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 10:48:11.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-07-24 14:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81ee6b75f9a90fd94c4f68eb8d2ca7f849885e75","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 22:25:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdfb4205acc29f7893df4cdc55ec1dc885295d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 01:17:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ee405c557b4d2594b1d1209d358196 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/332b70c21c3c4b3389bdc0c2eda321ac : SUCCESS in 39m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/906329c458174b5283c6ca7d44390e8b : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"51ba74be82c9058ed330136bd7050b646db0843f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-21 11:47:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5d69165585699f13c24259dae114c3180a4063d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-21 11:48:14.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, thanks for automating this Sean!","accounts_in_message":[],"_revision_number":2},{"id":"a7277d9e7095a94c49a11fec9939a55bca8cc6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:18:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06465543c104f6cb20d7dc65f699b86 : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09bf444d3a5a48feba25f395b8ebb1d6 : SUCCESS in 28m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85ad7162cc7b42d38ced5dfde3c4628f : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5d6a7a78f7d4553863c3acb09e13b63336eee78d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-21 17:36:27.000000000","message":"Patch Set 2:\n\nActually this looks OK. Maybe one thing: meanwhile a \u0027high\u0027 importance bugfix got merged [1] (and one \u0027low\u0027 importance bugfix is on the gate). Should we include those, or simply release this one and we will schedule another release later on? What do you think?\n\n[1] https://review.opendev.org/#/c/738199/","accounts_in_message":[],"_revision_number":2},{"id":"d24e0f19d41df9042939391a6f87b541af7dc4a2","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-23 16:36:08.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e Actually this looks OK. Maybe one thing: meanwhile a \u0027high\u0027\n \u003e importance bugfix got merged [1] (and one \u0027low\u0027 importance bugfix\n \u003e is on the gate). Should we include those, or simply release this\n \u003e one and we will schedule another release later on? What do you\n \u003e think?\n \u003e \n \u003e [1] https://review.opendev.org/#/c/738199/\n\nACK I can update this now.","accounts_in_message":[],"_revision_number":2},{"id":"25b625d3369f31fa544297f5bc796d4b4e56e0b1","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-23 16:38:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f152dcbda241e753d095ab59ada84845fe2860e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-23 16:54:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af182683d21b666ee4eacf1b008fe161c18a039f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-23 16:58:11.000000000","message":"Patch Set 4:\n\n\u003e \u003e Actually this looks OK. Maybe one thing: meanwhile a \u0027high\u0027\n \u003e \u003e importance bugfix got merged [1] (and one \u0027low\u0027 importance bugfix\n \u003e \u003e is on the gate). Should we include those, or simply release this\n \u003e \u003e one and we will schedule another release later on? What do you\n \u003e \u003e think?\n \u003e \u003e\n \u003e \u003e [1] https://review.opendev.org/#/c/738199/\n \u003e \n \u003e ACK I can update this now.\n\nThanks Lee for the update! (I guess the latest patch was accidentally missed, so I\u0027ve added that. o:))\n\nI still don\u0027t see any patch that would require minor version bump, but correct me if I\u0027m wrong.","accounts_in_message":[],"_revision_number":4},{"id":"d8898433834d2e549b5a4ed50f9e397287cb1ba2","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-23 17:05:05.000000000","message":"Patch Set 4: Code-Review+1\n\nOps, missed 5482dde329 thanks for correcting Elod!","accounts_in_message":[],"_revision_number":4},{"id":"575ca69376346df4b32881802d383b1db6f29a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:43:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39da447ef864b4eb123251eca1f2e89 : SUCCESS in 7m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81c39c76439f424f8cfee386ea0456e5 : SUCCESS in 41m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f8212151ec1e47f7b37a016a334c199b : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":4},{"id":"dac2b8f690be316863b7e6681082eef32331a0cc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-24 07:54:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00557f6f8455c80cc0ad1d0f8bdd27b39505e81c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-24 07:59:32.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34d8bc2a58041a74b112daee5fbf1ca27cbea4c0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-24 08:06:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3552d66016b5c5390f88698103e53c749761ad9f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-24 08:10:44.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks Gibi for the review, I\u0027ve updated the patch accordingly!","accounts_in_message":[],"_revision_number":5},{"id":"5a0f37bde84b1e09606423b84222a90bb1ffb78f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-24 08:13:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e8c7d53c32b909deadbc9c9b15074df3f2eb7760","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:16:51.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"e43dc7d455beea36cbf9001312f01c55ab82bfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:21:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400340d6f3c44b12b6101752b55b73b8 : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f434a3066ed4086bb49393cafba1e27 : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8e1d3822628944acb13314eafd5290a6 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":5},{"id":"3376155cf59e57e77d7cc511b6fce7ef18bd64db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 09:01:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a8d9cb664549f523ab016ec03f7def3e8f68e1d3","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-24 10:48:11.000000000","message":"Patch Set 5: Code-Review+1\n\nIt\u0027s marginal but I\u0027m not going to argue about trivial minor version release bumps.","accounts_in_message":[],"_revision_number":5},{"id":"f264d1890d9a0f27e32e5cbf1cb21808e8b04a6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 14:36:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"54539a1f5f9cb326369199afa55539e83d599618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:36:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3be4e9e5b541943e955da67b38e8b470a8d47cda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:50:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0aa0d3fe13852d06e5b61bb9f9070489ab2731e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:50:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656439313ada4e8db84c747a28e7fccb : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b375c9ef16d484d95cb74dafe0282f2 : SUCCESS in 8m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a9465369dee425cab786e3ad35bef2c : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d0f2d6de509faabd124d2a18c9c61d7323a7c525","revisions":{"8ddd72715fc107313b5242ae515b31994fc73f26":{"kind":"REWORK","_number":1,"created":"2020-07-17 22:25:09.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/60/741760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/741760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/741760/1"}}},"commit":{"parents":[{"commit":"805b89697562427a3d40fc94a51a88df23054055","subject":"Release openstackdocstheme 2.2.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/805b89697562427a3d40fc94a51a88df23054055"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"subject":"Release nova for stable/stein","message":"Release nova for stable/stein\n\nThis release picks up new commits to nova since\nthe last release from stable/stein.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nThis is not a release team activity, it is just to help out with\nstable releases. If there is no response to the patch, we will\nassume a release is not needed and abandon the patch after one\nweek.\n\nChange-Id: I3d6e9cc1aaafa640bdd884278d0b7e41d187074b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ddd72715fc107313b5242ae515b31994fc73f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ddd72715fc107313b5242ae515b31994fc73f26"}]},"branch":"refs/heads/master"},"438b35c58a05e3a3d3fe2d2c76e450bf93dfb906":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-07-21 11:47:33.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/60/741760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/741760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/741760/2"}}},"commit":{"parents":[{"commit":"25cb16ae9a510820775ca860c530cd71ff8ebea8","subject":"Merge \"Release octavia-lib 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25cb16ae9a510820775ca860c530cd71ff8ebea8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-07-21 11:47:23.000000000","tz":60},"subject":"nova: Release 19.2.1","message":"nova: Release 19.2.1\n\nNo minor version bump required with this release.\n\n$ git log --oneline --no-merges --no-decorate 19.2.0..HEAD\nce1287aa7f Guard against missing image cache directory\nc3dd9f86f1 Check cherry-pick hashes in pep8 tox target\n42e6320397 Revert \"Make greande jobs n-v for EM and oldest stable\"\n5429fd3d6e Add admin doc information about image cache resource accounting\n86be251451 Reserve DISK_GB resource for the image cache\nb05bf09b2d Make greande jobs n-v for EM and oldest stable\n21642202ff libvirt: avoid cpu check at s390x arch\na7d8aa6997 libvirt: Don\u0027t delete disks on shared storage during evacuate\nd7a1cb5724 Add functional test for bug 1550919\n86328d1146 fix scsi disk unit number of the attaching volume when cdrom bus is scsi\n4413e9bf56 Make quotas respect instance_list_per_project_cells\ne1116ee3b7 Update scheduler instance info at confirm resize\n2a15e0096b Reproduce bug 1869050\na63c97fd2d Revert \"nova shared storage: rbd is always shared storage\"\n1d66884af7 Add retry to cinder API calls related to volume detach\n\nChange-Id: I3d6e9cc1aaafa640bdd884278d0b7e41d187074b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/438b35c58a05e3a3d3fe2d2c76e450bf93dfb906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/438b35c58a05e3a3d3fe2d2c76e450bf93dfb906"}]},"branch":"refs/heads/master"},"6ef2415390d8ae0753bbec1dc75ec96179720219":{"kind":"REWORK","_number":3,"created":"2020-07-23 16:38:13.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/60/741760/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/741760/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/741760/3"}}},"commit":{"parents":[{"commit":"25cb16ae9a510820775ca860c530cd71ff8ebea8","subject":"Merge \"Release octavia-lib 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25cb16ae9a510820775ca860c530cd71ff8ebea8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-07-23 16:37:10.000000000","tz":60},"subject":"nova: Release 19.2.1","message":"nova: Release 19.2.1\n\nNo minor version bump required with this release.\n\n7b005f3785 zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\nce1287aa7f Guard against missing image cache directory\neaecf7e737 Remove stale nested backport from InstancePCIRequests\nc3dd9f86f1 Check cherry-pick hashes in pep8 tox target\n42e6320397 Revert \"Make greande jobs n-v for EM and oldest stable\"\n5429fd3d6e Add admin doc information about image cache resource accounting\n86be251451 Reserve DISK_GB resource for the image cache\nb05bf09b2d Make greande jobs n-v for EM and oldest stable\n7ff71e6b87 Create instance action when burying in cell0\n21642202ff libvirt: avoid cpu check at s390x arch\na7d8aa6997 libvirt: Don\u0027t delete disks on shared storage during evacuate\nd7a1cb5724 Add functional test for bug 1550919\n86328d1146 fix scsi disk unit number of the attaching volume when cdrom bus is scsi\n4413e9bf56 Make quotas respect instance_list_per_project_cells\ne1116ee3b7 Update scheduler instance info at confirm resize\n2a15e0096b Reproduce bug 1869050\na63c97fd2d Revert \"nova shared storage: rbd is always shared storage\"\n1d66884af7 Add retry to cinder API calls related to volume detach\n\nChange-Id: I3d6e9cc1aaafa640bdd884278d0b7e41d187074b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ef2415390d8ae0753bbec1dc75ec96179720219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ef2415390d8ae0753bbec1dc75ec96179720219"}]},"branch":"refs/heads/master"},"1042d1727b55eaf02a428c5cc19e497a34da8632":{"kind":"REWORK","_number":4,"created":"2020-07-23 16:54:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/60/741760/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/741760/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/741760/4"}}},"commit":{"parents":[{"commit":"25cb16ae9a510820775ca860c530cd71ff8ebea8","subject":"Merge \"Release octavia-lib 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25cb16ae9a510820775ca860c530cd71ff8ebea8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-07-23 16:53:46.000000000","tz":120},"subject":"nova: Release 19.2.1","message":"nova: Release 19.2.1\n\nNo minor version bump required with this release.\n\n5482dde329 libvirt: Mark e1000e VIF as supported\n7b005f3785 zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\nce1287aa7f Guard against missing image cache directory\neaecf7e737 Remove stale nested backport from InstancePCIRequests\nc3dd9f86f1 Check cherry-pick hashes in pep8 tox target\n42e6320397 Revert \"Make greande jobs n-v for EM and oldest stable\"\n5429fd3d6e Add admin doc information about image cache resource accounting\n86be251451 Reserve DISK_GB resource for the image cache\nb05bf09b2d Make greande jobs n-v for EM and oldest stable\n7ff71e6b87 Create instance action when burying in cell0\n21642202ff libvirt: avoid cpu check at s390x arch\na7d8aa6997 libvirt: Don\u0027t delete disks on shared storage during evacuate\nd7a1cb5724 Add functional test for bug 1550919\n86328d1146 fix scsi disk unit number of the attaching volume when cdrom bus is scsi\n4413e9bf56 Make quotas respect instance_list_per_project_cells\ne1116ee3b7 Update scheduler instance info at confirm resize\n2a15e0096b Reproduce bug 1869050\na63c97fd2d Revert \"nova shared storage: rbd is always shared storage\"\n1d66884af7 Add retry to cinder API calls related to volume detach\n\nChange-Id: I3d6e9cc1aaafa640bdd884278d0b7e41d187074b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1042d1727b55eaf02a428c5cc19e497a34da8632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1042d1727b55eaf02a428c5cc19e497a34da8632"}]},"branch":"refs/heads/master"},"d0f2d6de509faabd124d2a18c9c61d7323a7c525":{"kind":"REWORK","_number":5,"created":"2020-07-24 08:06:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/60/741760/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/741760/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/741760/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/741760/5"}}},"commit":{"parents":[{"commit":"25cb16ae9a510820775ca860c530cd71ff8ebea8","subject":"Merge \"Release octavia-lib 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25cb16ae9a510820775ca860c530cd71ff8ebea8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-17 22:25:07.000000000","tz":-300},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-07-24 08:03:26.000000000","tz":120},"subject":"nova: Release 19.3.0","message":"nova: Release 19.3.0\n\nMinor version bump required with this release due to patch\n\u0027libvirt: Mark e1000e VIF as supported\u0027.\n\n5482dde329 libvirt: Mark e1000e VIF as supported\n04a8399f1d compute: Allow snapshots to be created from PAUSED volume backed instances\n7b005f3785 zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\nce1287aa7f Guard against missing image cache directory\neaecf7e737 Remove stale nested backport from InstancePCIRequests\nc3dd9f86f1 Check cherry-pick hashes in pep8 tox target\n42e6320397 Revert \"Make greande jobs n-v for EM and oldest stable\"\n5429fd3d6e Add admin doc information about image cache resource accounting\n86be251451 Reserve DISK_GB resource for the image cache\nb05bf09b2d Make greande jobs n-v for EM and oldest stable\n7ff71e6b87 Create instance action when burying in cell0\n21642202ff libvirt: avoid cpu check at s390x arch\na7d8aa6997 libvirt: Don\u0027t delete disks on shared storage during evacuate\nd7a1cb5724 Add functional test for bug 1550919\n86328d1146 fix scsi disk unit number of the attaching volume when cdrom bus is scsi\n4413e9bf56 Make quotas respect instance_list_per_project_cells\ne1116ee3b7 Update scheduler instance info at confirm resize\n2a15e0096b Reproduce bug 1869050\na63c97fd2d Revert \"nova shared storage: rbd is always shared storage\"\n1d66884af7 Add retry to cinder API calls related to volume detach\n\nChange-Id: I3d6e9cc1aaafa640bdd884278d0b7e41d187074b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0f2d6de509faabd124d2a18c9c61d7323a7c525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0f2d6de509faabd124d2a18c9c61d7323a7c525"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
