)]}'
{"id":"openstack%2Freleases~742250","triplet_id":"openstack%2Freleases~master~I1a4a2f926329f16aaadb44ba7af5ddb38360b9dc","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I1a4a2f926329f16aaadb44ba7af5ddb38360b9dc","subject":"Add ussuri release note links for osc-lib and python-openstackclient","status":"MERGED","created":"2020-07-21 16:56:06.000000000","updated":"2020-07-21 17:48:09.000000000","submitted":"2020-07-21 17:48:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742250-1595353689906-dfe663bb","meta_rev_id":"b2efeb79ea489bf5091c3f02c891c4bfba6ef42d","_number":742250,"virtual_id_number":742250,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 17:48:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-21 17:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-21 17:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-21 17:20:07.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 17:20:07.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-21 17:34:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-21 17:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7c5bac0e52fd034100a70658aa6b0b3e02586ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 16:56:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23da4a1fc8becbc35f219a26c1e7920e9c9dabb5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 16:59:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba2f4368d5b6e230f30e3a21f8f249e2b7ef57d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:08:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013fafd5d08f456fb40c934fe0720b3d : FAILURE in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7980ec91ccd64456aeccb0bee56f10bb : FAILURE in 4m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ded0c14223e444c68c48aebf7d9a3481 : FAILURE in 4m 53s\n\nWarning:\n  Job openstack-tox-validate: unable to map line for file comments: \n  stderr: \u0027fatal: file deliverables/ussuri/osc-lib.yaml has only 21 lines\u0027","accounts_in_message":[],"_revision_number":1},{"id":"d0c1cfcf56c31ecd0b63fe50468aed45bc9d15cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 17:11:45.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1ac0683cfcc4b126345b4acbcc9ec8f293d3853","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 17:16:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69607f0f83756c7b89a02604d93e2222c8190e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:29:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c440602deedb42e3ba75f917d3a6b1e7 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2575a5e291d14b8ab08ce9c8a107f9da : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d1153f2a13f44d5fb7e286fc8e902980 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e6659aec8c1cb3637cef25e6bc376f54e37eb7c0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 17:34:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5de54b8da4470d49a1594ab71c5ae4f5d6d5e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:38:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ec4b9c81e0c7f50823396d8c8fdb9189118fc4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:48:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b2efeb79ea489bf5091c3f02c891c4bfba6ef42d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:48:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62219d8e38dd4d95b52b6a63b398f86b : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf86c7af28dd457b931c7c6f7b98b0b2 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/338ef70ae00b44a08f46f21834ec4dd8 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc484b3e9b7b237f32246f0cfefb10c95f1d3681","revisions":{"92c09ff5565d17ba2c6f832ebdf940b4838a631c":{"kind":"REWORK","_number":1,"created":"2020-07-21 16:56:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/742250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/742250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/742250/1"}}},"commit":{"parents":[{"commit":"f32cffad0adbdf49f68459054ba0109cc4a58d3f","subject":"tooz 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f32cffad0adbdf49f68459054ba0109cc4a58d3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-21 16:55:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-21 16:55:58.000000000","tz":-420},"subject":"Add ussuri release note links for osc-lib and python-openstackclient","message":"Add ussuri release note links for osc-lib and python-openstackclient\n\nChange-Id: I1a4a2f926329f16aaadb44ba7af5ddb38360b9dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92c09ff5565d17ba2c6f832ebdf940b4838a631c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92c09ff5565d17ba2c6f832ebdf940b4838a631c"}]},"branch":"refs/heads/master"},"bc484b3e9b7b237f32246f0cfefb10c95f1d3681":{"kind":"REWORK","_number":2,"created":"2020-07-21 17:16:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/742250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/742250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/742250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/742250/2"}}},"commit":{"parents":[{"commit":"f32cffad0adbdf49f68459054ba0109cc4a58d3f","subject":"tooz 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f32cffad0adbdf49f68459054ba0109cc4a58d3f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-21 16:55:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-21 17:16:13.000000000","tz":0},"subject":"Add ussuri release note links for osc-lib and python-openstackclient","message":"Add ussuri release note links for osc-lib and python-openstackclient\n\nChange-Id: I1a4a2f926329f16aaadb44ba7af5ddb38360b9dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc484b3e9b7b237f32246f0cfefb10c95f1d3681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc484b3e9b7b237f32246f0cfefb10c95f1d3681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
