)]}'
{"id":"openstack%2Freleases~742478","triplet_id":"openstack%2Freleases~master~I79f061f57f04c155cf99ba80d8d96c935521e75d","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I79f061f57f04c155cf99ba80d8d96c935521e75d","subject":"minor update release for tripleo ussuri","status":"MERGED","created":"2020-07-22 16:22:31.000000000","updated":"2020-07-24 18:28:59.000000000","submitted":"2020-07-24 18:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742478-1595615339265-5004bf0a","meta_rev_id":"8233997c27dd363fa6807a39630ed2a0a480e876","_number":742478,"virtual_id_number":742478,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 18:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 18:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-07-24 15:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-07-24 15:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-24 18:15:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-24 15:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 21:19:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-24 15:08:15.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-07-24 15:24:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 18:15:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 18:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c913deef3fe2d83d3dadd96e1df87c3477bfb34f","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 16:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6110ccb7d850cfd9b3f581e09e7293eb179174","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 16:22:45.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"34bb380a13c3686baf2e61666cb346f5b34a360f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:23:51.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"87bbed2c8ca1b5ebce3683027e0b25e313a237ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:38:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6691b5f7d2bc4ffc85fd182bddbd57d0 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b7ca6a6569542b8b812244cda40e1e3 : FAILURE in 14m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9dab20f72c6146b2b6252ca1715e9a9c : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"133e113af1e054b266c2ac6e9ef0cd1a22477f15","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-23 19:47:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fda4e39593698ae2aafb7b68082ed3a51a4e67a0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:49:23.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"0a01123341857b46879a8640066481db1f1868d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:12:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40657590d4734ae9979684dfac6a1600 : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e99c2e3bb780400bb2566fc0adb6b0e4 : FAILURE in 17m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f378cb97346048c0b0a069477046a613 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":2},{"id":"80f18ffd402b03f7187496da1de9bea57595954a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-23 20:14:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"86649f6bd26a61daa34d9da8af1f0f969afa945a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-23 20:14:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d71a44b05e29bf2348982bdab1ff9f4f5021e7cd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:15:13.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"9b3d41eadfb70e6bbca45abaffad4557275db409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:26:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686c8e92ea034642a8fda55c90c8eb86 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/605f889d6f01434e97ac270c7a227b92 : SUCCESS in 11m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1dd4485805ae48bbb6146f6788bd8d4a : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"34423bb511d6db151e772cddccda9445c8ee02e0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-23 21:19:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bc688e2bff39a5d4467a44d193958b298454073b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-24 04:59:41.000000000","message":"Patch Set 3:\n\nI see some more fixes merged recently in stable/ussuri, can we get those too?\ni mainly looked at tripleo-ansible:- https://github.com/openstack/tripleo-ansible/commit/6ac7cc478d6e1a0f63a2caf2d9e1340c28363c04 which will be needed for ansible-2.9.11, but good to take latest for all other projects","accounts_in_message":[],"_revision_number":3},{"id":"7d122281d22e93ad5d0db8454c323ed21b35a177","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 09:56:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nHello","accounts_in_message":[],"_revision_number":3},{"id":"6971ce2a4130efaf94685265d7bd5f2b09610859","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 10:04:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a6f76f57c290ca3993c4df586fd5b64b45bfce7e","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:53:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52ccc8d30538b80bf842c689b2cc420f9f416500","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:53:51.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"2072bd345b6b56aeee45bea11e16d3e10bbdf593","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:54:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"089df1a38b0af136d58d64605ec6b0d966ba4a55","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:54:51.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"c1071167fb6adc239f4648d3a32ade35413617f4","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:58:41.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"47aeab149e084593a91bd8168d62870c03828f65","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 15:05:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"da78a45ccf3cf6cdf2616976d2f2cc99c6ee2874","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 15:07:44.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"82bb1305b77f5576d69d30912c8fa77c4ddab5e6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-24 15:08:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9b9808475b4d2da55aceaf5b9aad85576f9fd002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 15:23:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613e72eeceef43b9a17c71e467c5856d : SUCCESS in 10m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5d184fbc42146dd8d54ec19dbe0d973 : SUCCESS in 11m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cc615204f8364e078377c8222f9d0faf : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":5},{"id":"8908ce3cc449d57a4b588d00c109e5d096d7ee29","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 15:24:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2956c753c728c486b106e624728db7df567085a0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:15:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dc043b2fda0a6956e001e9bf4e15fa506b76e860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:16:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fe81879c1ee607bf0a61d19f6eab8f3a56c2a9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:28:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b66ebaa2d314601bee7ab2ae78da055 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d48e6f74dc34c03ae0eccb11897f142 : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bea50e4b73bb44529be5c9aabf064a51 : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":5},{"id":"8233997c27dd363fa6807a39630ed2a0a480e876","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:28:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"882981c2d9bddfe87902eba68c88beae0b79aaf9","revisions":{"66f91c8a33b21fe879ec11e06e8ba8cbf6da9b0b":{"kind":"REWORK","_number":1,"created":"2020-07-22 16:22:31.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/78/742478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/742478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/742478/1"}}},"commit":{"parents":[{"commit":"e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9","subject":"Merge \"oslo.log 4.1.3 (ussuri)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:21:52.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:22:08.000000000","tz":-360},"subject":"minor update release for tripleo ussuri","message":"minor update release for tripleo ussuri\n\nDepends-On: https://review.opendev.org/742463\nChange-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66f91c8a33b21fe879ec11e06e8ba8cbf6da9b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66f91c8a33b21fe879ec11e06e8ba8cbf6da9b0b"}]},"branch":"refs/heads/master"},"33a238e223f9c5798f0d578db65da39df4733dfa":{"kind":"REWORK","_number":2,"created":"2020-07-23 19:47:46.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/78/742478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/742478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/742478/2"}}},"commit":{"parents":[{"commit":"e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9","subject":"Merge \"oslo.log 4.1.3 (ussuri)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:21:52.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-23 19:47:41.000000000","tz":-360},"subject":"minor update release for tripleo ussuri","message":"minor update release for tripleo ussuri\n\nDepends-On: https://review.opendev.org/742463\nChange-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33a238e223f9c5798f0d578db65da39df4733dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33a238e223f9c5798f0d578db65da39df4733dfa"}]},"branch":"refs/heads/master"},"8aa726b67f05452835fac7ce06586d0350a0b3c8":{"kind":"REWORK","_number":3,"created":"2020-07-23 20:14:02.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/78/742478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/742478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/742478/3"}}},"commit":{"parents":[{"commit":"e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9","subject":"Merge \"oslo.log 4.1.3 (ussuri)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:21:52.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-23 20:13:58.000000000","tz":-360},"subject":"minor update release for tripleo ussuri","message":"minor update release for tripleo ussuri\n\nDepends-On: https://review.opendev.org/742463\nChange-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8aa726b67f05452835fac7ce06586d0350a0b3c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8aa726b67f05452835fac7ce06586d0350a0b3c8"}]},"branch":"refs/heads/master"},"c8a721be676e209ac184ecbc7ea0e2a8e3641e05":{"kind":"REWORK","_number":4,"created":"2020-07-24 14:53:14.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/78/742478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/742478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/742478/4"}}},"commit":{"parents":[{"commit":"e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9","subject":"Merge \"oslo.log 4.1.3 (ussuri)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:21:52.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-24 14:53:08.000000000","tz":-360},"subject":"minor update release for tripleo ussuri","message":"minor update release for tripleo ussuri\n\nDepends-On: https://review.opendev.org/742463\nChange-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8a721be676e209ac184ecbc7ea0e2a8e3641e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8a721be676e209ac184ecbc7ea0e2a8e3641e05"}]},"branch":"refs/heads/master"},"882981c2d9bddfe87902eba68c88beae0b79aaf9":{"kind":"REWORK","_number":5,"created":"2020-07-24 15:05:38.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/78/742478/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/742478/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/742478/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/742478/5"}}},"commit":{"parents":[{"commit":"e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9","subject":"Merge \"oslo.log 4.1.3 (ussuri)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7eb3ce4b30d9530e9007bbabf13a2b547f98ee9"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 16:21:52.000000000","tz":-360},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2020-07-24 15:04:28.000000000","tz":0},"subject":"minor update release for tripleo ussuri","message":"minor update release for tripleo ussuri\n\nDepends-On: https://review.opendev.org/742463\nChange-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/882981c2d9bddfe87902eba68c88beae0b79aaf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/882981c2d9bddfe87902eba68c88beae0b79aaf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
