)]}'
{"id":"openstack%2Freleases~742485","triplet_id":"openstack%2Freleases~master~I31b4a265197329394404a6edc7027d78e35c66c5","project":"openstack/releases","branch":"master","topic":"oslo-master","hashtags":[],"change_id":"I31b4a265197329394404a6edc7027d78e35c66c5","subject":"Fix handling of unreleased repos","status":"ABANDONED","created":"2020-07-22 16:53:28.000000000","updated":"2020-07-23 14:49:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"121905ff48775ff5553d27c16d21ff227c9e4edd","_number":742485,"virtual_id_number":742485,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-22 19:08:52.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-07-23 08:56:28.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-22 17:17:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 17:17:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 19:08:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-23 08:56:28.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"70ba70de51dbaa7343adada69ded98ba7c7c4dcb","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-22 16:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afffa135b4ac9c3021f868fbd5e556bfd5831ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:17:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f942bab64b948f4a684932ceefc0dbd : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c9dda9735d44ea1938fb6ab05dbc111 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37b0bdf7b89e4590a49eff1034e1101b : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a976af4816c4150a4cacb13dff9827f : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df7726a3ebc3499798ea65105f3ad779 : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6cc07d31b4584d4eab048c890948ae17 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6ec8586ba42451abf8c466ecad2806f : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d62757542700141a74e3f66129495db04c11f2f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 19:08:52.000000000","message":"Patch Set 1:\n\nI think we want to go with https://review.opendev.org/#/c/742454","accounts_in_message":[],"_revision_number":1},{"id":"e1b8e234929bcbe74e3d6bfb26a098f431feb090","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-23 08:56:28.000000000","message":"Patch Set 1:\n\nYes that sounds redundant","accounts_in_message":[],"_revision_number":1},{"id":"869a33d04d98e19e11195be56dd1e4178db9dd0c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-23 14:49:22.000000000","message":"Abandoned\n\nSounds good. I\u0027ll rebase the subsequent patch on that one.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"587512275f8c999040cd1a0dd8504c589443c52d","revisions":{"587512275f8c999040cd1a0dd8504c589443c52d":{"kind":"REWORK","_number":1,"created":"2020-07-22 16:53:28.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/85/742485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/742485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/742485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/742485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/742485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/742485/1"}}},"commit":{"parents":[{"commit":"3b9d403a3216eaf57ed6bdd558887ac7bc63ff60","subject":"Oslo releases for 2020-07-22","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b9d403a3216eaf57ed6bdd558887ac7bc63ff60"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-07-22 15:52:20.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-07-22 16:53:10.000000000","tz":0},"subject":"Fix handling of unreleased repos","message":"Fix handling of unreleased repos\n\nWithout this change list_unreleased_changes explodes when it gets to\na repo that hasn\u0027t been released yet because the \u0027commits\u0027 value of\nthe data is None.\n\nAlternatively, we could remove the ability to configure the\n--ignore-not-yet-released so we don\u0027t have to handle these cases\nat all, but then those projects won\u0027t show up in the output.\n\nChange-Id: I31b4a265197329394404a6edc7027d78e35c66c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/587512275f8c999040cd1a0dd8504c589443c52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/587512275f8c999040cd1a0dd8504c589443c52d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
