)]}'
{"id":"openstack%2Freleases~742486","triplet_id":"openstack%2Freleases~master~I452aa5500bda1bf607ba11baa92be3377945a4ab","project":"openstack/releases","branch":"master","topic":"oslo-master-2","hashtags":[],"change_id":"I452aa5500bda1bf607ba11baa92be3377945a4ab","subject":"Add newline between projects in std output format","status":"MERGED","created":"2020-07-22 16:53:28.000000000","updated":"2020-07-24 18:31:58.000000000","submitted":"2020-07-24 18:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742486-1595615518192-46e369dc","meta_rev_id":"ae798952ea3cbad783f1e489cc2593cd9afe2227","_number":742486,"virtual_id_number":742486,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 18:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 18:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-24 17:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-07-24 17:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-24 18:19:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 08:55:35.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-24 09:04:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 18:19:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 18:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e280431b6923c076aee360cd28b31ac4669295db","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-22 16:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb20875154b023e072d6297f99e6fd56204bf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:21:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffedeeb69c0d49399245572b1bfce5ae : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b3100e92ac41a5b0705f6d9ea16bb4 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e45f346d62ba47f19e3d105553076975 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3eee188996404e99646eeee2d75fd4 : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/138befaa954641d1ae54796c8105730e : SUCCESS in 4m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5a9b263941cb403fa8d7de80b4919b8d : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5f81efe890343c4852190722b9f5bae : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7174177151d86eff4a99c697063ef0324ed4ef79","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-23 08:55:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad0af7d21cadef0ce0d4cd9bdb8c3f0187c90550","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 09:04:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI added colored outputs [1] to more easily show results between repos (when more than 1 repo is passed to the CLI), but anyway a new line can\u0027t hurt too \n\n[1] https://review.opendev.org/#/c/742454/","accounts_in_message":[],"_revision_number":1},{"id":"69006077746d0e4a0ecfe68d5df1105dad1ef834","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 17:13:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0e6a940a681313b96c8b5b036d307564680b77c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7169398e9a5c4820a8e71cb9aff8a374 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c405ffdfd3fd4b308b4275baca397b4c : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efe8aca667414c0fa42584e043237707 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0ff49dd1a641e79ac1bd33fe9a7662 : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0097a559ae034396bc4b96a5fd9f1f13 : SUCCESS in 6m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ef3a954c43f442f5bc9a8a856f590977 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2684d89205434f759c91a958d87496be : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a56f34712b8dc11fc2bc1a67ed9975c57b0203e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:19:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"efd9b7426b94cf4d4ba6116bd802be1f196cd087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:19:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22400b0a592949a5bf61995348e57e14fa419e4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:31:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ad8454d5ca4c7fa52637e93ffdeb58 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b07c669c3b24eada6f3c40013bbedd5 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aea441268f63470ba929971b1126c882 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9111de2ddf3f460d93989c87e72fef1c : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81c3e6e91bb04f25880a023b3fc9a5d9 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/067ffad028c44a1ca7912c61bcc816cd : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e12da5efa630a2189dba0838870690239b22bbe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:31:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0bd6e76f35846a6cb907fd1dc183834e2ce23fc","revisions":{"6fcf3be913e4115bed0841395550583531538159":{"kind":"REWORK","_number":1,"created":"2020-07-22 16:53:28.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/86/742486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/742486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/742486/1"}}},"commit":{"parents":[{"commit":"587512275f8c999040cd1a0dd8504c589443c52d","subject":"Fix handling of unreleased repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/587512275f8c999040cd1a0dd8504c589443c52d"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-07-22 16:08:10.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-07-22 16:53:10.000000000","tz":0},"subject":"Add newline between projects in std output format","message":"Add newline between projects in std output format\n\nWithout this it is very difficult to visually distinguish between\nthe project outputs.\n\nChange-Id: I452aa5500bda1bf607ba11baa92be3377945a4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6fcf3be913e4115bed0841395550583531538159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6fcf3be913e4115bed0841395550583531538159"}]},"branch":"refs/heads/master"},"d0bd6e76f35846a6cb907fd1dc183834e2ce23fc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-24 17:13:20.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/742486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/742486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/742486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/742486/2"}}},"commit":{"parents":[{"commit":"4431125d1055173734a3a5835ff822960b48c810","subject":"Merge \"Release monasca-ui for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4431125d1055173734a3a5835ff822960b48c810"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-07-22 16:08:10.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-24 17:12:45.000000000","tz":120},"subject":"Add newline between projects in std output format","message":"Add newline between projects in std output format\n\nWithout this it is very difficult to visually distinguish between\nthe project outputs.\n\nChange-Id: I452aa5500bda1bf607ba11baa92be3377945a4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0bd6e76f35846a6cb907fd1dc183834e2ce23fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0bd6e76f35846a6cb907fd1dc183834e2ce23fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
