)]}'
{"id":"openstack%2Freleases~742492","triplet_id":"openstack%2Freleases~master~I65232a3102b8a061006524b7f9b967dfbb38e8f6","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I65232a3102b8a061006524b7f9b967dfbb38e8f6","subject":"tripleo victoria-1 release","status":"MERGED","created":"2020-07-22 17:02:31.000000000","updated":"2020-07-27 05:55:49.000000000","submitted":"2020-07-27 05:55:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742492-1595829349272-46e369dc","meta_rev_id":"88b664a6efaa33db1970b5c17a83354652cfb80f","_number":742492,"virtual_id_number":742492,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 05:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 18:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-07-27 00:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-07-27 05:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-07-27 05:43:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-24 15:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 21:14:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-07-23 21:20:00.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-24 18:17:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-27 05:43:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-27 05:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a5140d2044a8d0b23b6623c05084e9bd82f72bb","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 17:02:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57829df6d325d7e50064c40ace6c3cb34639fb31","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 17:02:39.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"cccefa25d9a0d44c9e24c365bf14ddcea3ca9c50","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:03:59.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c72eff6f7286b8c1357c242d4c88af00917aeda1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:30:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4c073ae36a49db9f82944c76502da6 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/247dd6c8f7b0478ca4eaaa14bf4feac0 : FAILURE in 7m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8efccad6bbad4f5b89e0e8394830f501 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d6f45c92585c1f5cdf5819038c7a0013f44ab20e","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-23 20:36:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c53a29ba71f090319103b379dadd178d48297a75","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:37:49.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3315a72b9593514bbc0a682a5ba6be18fd3887d7","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-23 20:42:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4b15f51d1b00266449c310819ab6983374111533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:45:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc414a76a13e4c58ba7a5579e8715527 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75e6d4cdb40c4038b11bf7e09138813d : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9e4230b17dab48a2aeb5d60acf570f04 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"12ebd6ab37f244cb8793b424268e72ff1dce205c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-23 21:14:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da56144b1f1b2922605e9edb60c12d3833a38c52","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-23 21:20:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5b95621bbf3ccd98f238b84fa07f8086144366a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 09:12:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62bf6d706c025739ecfb5c686d362c9d67f0015b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 09:56:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nHello","accounts_in_message":[],"_revision_number":2},{"id":"36da27d2a32b814cf16c43f5fcd96b4d14cc09c3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 10:05:39.000000000","message":"Patch Set 2:\n\nPlease see my comments on your ussuri patch [1] to sync your version here.\n\nThanks\n\n[1] https://review.opendev.org/#/c/742478/3","accounts_in_message":[],"_revision_number":2},{"id":"4da2e14f3ec94c695fc762daab68506a56018850","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:51:49.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6537c900767211ebdf71d550cdc8eb07eaa7099c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 14:57:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"25272860283366b19c5b38dfba09c397b5483f6e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 15:00:15.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"124663254acaf8a31dc94769470cc5b310f483b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 15:12:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7686cd311b4743bd8ed2bd346ad4b4bb : SUCCESS in 8m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf17acdcf3a649ed9ce9667d08101305 : SUCCESS in 7m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bdfffb9f99084c8092f080c4baf1079e : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"69882aaa496011fd8cf41f18b71b8e977222cfea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 15:29:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nHello,\n\nThanks for the changes, see my inline comment:","accounts_in_message":[],"_revision_number":3},{"id":"61691526c1bee3b16c7e6526ac3dd700f23d2cac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 15:35:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e2e9cab159f47158973201f1667300594cf94c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 16:56:38.000000000","message":"Patch Set 3:\n\nOnly +1 to ensure that tripleo for ussuri [1] is merged first (cf. my previous comments), I\u0027ll update my vote when ussuri will be merged.\n\n[1] https://review.opendev.org/#/c/742478/","accounts_in_message":[],"_revision_number":3},{"id":"993b615abbacdb4c1d71b894ccc5b29679669a19","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:17:53.000000000","message":"Patch Set 3: Code-Review+2\n\n\u003e Only +1 to ensure that tripleo for ussuri [1] is merged first (cf.\n \u003e my previous comments), I\u0027ll update my vote when ussuri will be\n \u003e merged.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/742478/\n\nI don\u0027t think it\u0027s necessary to depend on that patch being released first.","accounts_in_message":[],"_revision_number":3},{"id":"1ff3266d5f20b22b3aa143554896fc57f49ddbb0","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-27 00:12:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"96eb7bb28483218f2361e9d171248d7de0960024","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-27 05:43:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bcb1cc82ca110081588e0e5aeb5028d81e333c34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 05:44:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"91722240cac809821918fef96118f47fc9b35549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 05:55:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e0d98deabd422f83e24e647b5115c5 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979ae01d4bfa4f87a84e28fdab74f3d7 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0bce78f00d604b3f9ee64c176e8da0d1 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"88b664a6efaa33db1970b5c17a83354652cfb80f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 05:55:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ff19cd4c632893f0875b2ee8b33a9a7e37a3ec7a","revisions":{"4a851b1958fe484a0c71964528345615c8c7adcf":{"kind":"REWORK","_number":1,"created":"2020-07-22 17:02:31.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/92/742492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/742492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/742492/1"}}},"commit":{"parents":[{"commit":"a7d862c8f00b3e6b5b820854cc3958993eb4e592","subject":"Merge \"Add new release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7d862c8f00b3e6b5b820854cc3958993eb4e592"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 17:01:54.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 17:01:57.000000000","tz":-360},"subject":"tripleo victoria-1 release","message":"tripleo victoria-1 release\n\nDepends-On: https://review.opendev.org/#/c/742432\nChange-Id: I65232a3102b8a061006524b7f9b967dfbb38e8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a851b1958fe484a0c71964528345615c8c7adcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a851b1958fe484a0c71964528345615c8c7adcf"}]},"branch":"refs/heads/master"},"5ea07a4bd94e6c154e923f2f7f599633c9168a19":{"kind":"REWORK","_number":2,"created":"2020-07-23 20:36:50.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/92/742492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/742492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/742492/2"}}},"commit":{"parents":[{"commit":"a7d862c8f00b3e6b5b820854cc3958993eb4e592","subject":"Merge \"Add new release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7d862c8f00b3e6b5b820854cc3958993eb4e592"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 17:01:54.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-23 20:36:41.000000000","tz":-360},"subject":"tripleo victoria-1 release","message":"tripleo victoria-1 release\n\nDepends-On: https://review.opendev.org/#/c/742432\nChange-Id: I65232a3102b8a061006524b7f9b967dfbb38e8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ea07a4bd94e6c154e923f2f7f599633c9168a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ea07a4bd94e6c154e923f2f7f599633c9168a19"}]},"branch":"refs/heads/master"},"ff19cd4c632893f0875b2ee8b33a9a7e37a3ec7a":{"kind":"REWORK","_number":3,"created":"2020-07-24 14:57:22.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/92/742492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/742492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/742492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/742492/3"}}},"commit":{"parents":[{"commit":"a7d862c8f00b3e6b5b820854cc3958993eb4e592","subject":"Merge \"Add new release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7d862c8f00b3e6b5b820854cc3958993eb4e592"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-07-22 17:01:54.000000000","tz":-360},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2020-07-24 14:57:19.000000000","tz":0},"subject":"tripleo victoria-1 release","message":"tripleo victoria-1 release\n\nDepends-On: https://review.opendev.org/#/c/742432\nChange-Id: I65232a3102b8a061006524b7f9b967dfbb38e8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff19cd4c632893f0875b2ee8b33a9a7e37a3ec7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff19cd4c632893f0875b2ee8b33a9a7e37a3ec7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
