)]}'
{"id":"openstack%2Freleases~742523","triplet_id":"openstack%2Freleases~master~If266ab3efa5bdde3642d13136363573208b1e4d6","project":"openstack/releases","branch":"master","topic":"cinder-eol-pike","hashtags":[],"change_id":"If266ab3efa5bdde3642d13136363573208b1e4d6","subject":"Cinder: EOL Pike","status":"MERGED","created":"2020-07-22 19:31:38.000000000","updated":"2020-07-29 21:26:51.000000000","submitted":"2020-07-29 21:26:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742523-1596058011670-be75a880","meta_rev_id":"e5dc6da7a66a7651079d6431095496a90480d81c","_number":742523,"virtual_id_number":742523,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 21:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-29 21:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-07-29 11:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-29 14:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-29 21:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-07-29 11:30:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 11:32:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-29 14:20:35.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-29 21:12:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-29 21:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2255be2f2d58f012ee3df60b2de34cce2a06f6fd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-22 19:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4276413932b7252d38d1b7d600a539f55e56fb70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:33:17.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c8931dd5a1ed9a36c808f579dd01d66053f995a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:41:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fa52f2262d481b8b4b2fc65ac34879 : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da3a3fb0f6084e6c8ac011be5ca7bd11 : SUCCESS in 7m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1286fd25cb4e4b3fb01789833ac84ca1 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3989e528f7ae5f3ce36ba0719fbd10cbd470d7f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 19:43:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"abf9b1f05d4abce81ad9eee9df2d133d4cdf12dc","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-23 13:35:16.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for proposing this.  Looks good.","accounts_in_message":[],"_revision_number":1},{"id":"7abe9d27bbe7155dfa2baf86fbe2bcb32942712f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-23 18:28:38.000000000","message":"Patch Set 1: Workflow-1\n\nGoing to hold this for a day or two.","accounts_in_message":[],"_revision_number":1},{"id":"4bc186a78abd9a10ec012fd9f41b68768e515d26","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-23 19:45:21.000000000","message":"Patch Set 1:\n\nHere\u0027s the reason for the hold: http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016076.html","accounts_in_message":[],"_revision_number":1},{"id":"e889093c7c12848b1d354b1bcc09ad6cc59e044d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-28 14:03:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19dd4c9a85cda63b6ab8d33bc433c5cf241032df","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:10:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"137cf41b9ad1303c5cce8ba4c06aa4b80b0196c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:22:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0100ee3cb3466197e9e0ef2bcd5fbc : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e849fdda18b8460ebaa42cc1214a5490 : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7783aa92fb64079b72e0a8b287af0ce : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f5e20070a3d554880ee6e8f8a303adf18b570ea8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 08:44:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cba76c3618aafdec97a2f96428ab14d74ae7263a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-29 11:25:46.000000000","message":"Patch Set 2: Workflow-1\n\nThanks, Hervé ... i don\u0027t know how i missed the brick update.","accounts_in_message":[],"_revision_number":2},{"id":"5538bb6d4fef3074d74980fa1d24c21050600635","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-29 11:29:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54e6b38a3afc1d0e13e87331983354c2bfb2320c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:30:12.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"8b38d22d8534a8c2f1fb2f4b22972e22b892d46b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 11:32:41.000000000","message":"Patch Set 3: Code-Review+2\n\n\u003e Thanks, Hervé ... i don\u0027t know how i missed the brick update.\n\nno problem :)","accounts_in_message":[],"_revision_number":3},{"id":"71777f3c4b1edcc86b28dc03bcb4a86be45cf407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:43:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529d8a8dcacc48f5804c0f0194701e10 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c956497199c04ca1a7814522d1e8757c : SUCCESS in 4m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d85752ed0044e27936d1e10fad000e8 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"cc174ee751d74d3fdb6e13a440bc7fb935d381ec","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-29 14:20:35.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good.  Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"e36383ae107eafa7b9e5c55b94cbdb47ea44a55e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 21:12:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8863051231f5afe525c58af0c9c79101633fd7f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 21:13:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c520c331c4fd60a6ddc9de6a497303a60b4d7896","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 21:26:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec738c21d108411aa357fb39e5d81c01 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4776fa7d7faa42b882f0703a39c143b7 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3884a9aa67c4ebb98bcc7b6835c3137 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c31b1f0f113957adb4165ec1c7410062520c584f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 21:26:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb6015c8e48a5f9489c4f46a893e386d88b1e71a","revisions":{"6348af456d23f2361dcaab543d2de0daf1b62e8c":{"kind":"REWORK","_number":1,"created":"2020-07-22 19:31:38.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/23/742523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/742523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/742523/1"}}},"commit":{"parents":[{"commit":"a7d862c8f00b3e6b5b820854cc3958993eb4e592","subject":"Merge \"Add new release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7d862c8f00b3e6b5b820854cc3958993eb4e592"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-22 19:29:46.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-22 19:29:46.000000000","tz":-240},"subject":"Cinder: EOL Pike","message":"Cinder: EOL Pike\n\nAs announced on the mailing list [0,1], the Cinder project team wishes\nto transition the stable/pike branches of its deliverables from\n\u0027unmaintained\u0027 to EOL, and hereby requests that pike-eol tags be\napplied to the stable/pike branches and the branches be deleted.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015838.html\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016050.html\n\nChange-Id: If266ab3efa5bdde3642d13136363573208b1e4d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6348af456d23f2361dcaab543d2de0daf1b62e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6348af456d23f2361dcaab543d2de0daf1b62e8c"}]},"branch":"refs/heads/master"},"994d3f174914ff8747a3847aec6d579c2d2eb8aa":{"kind":"REWORK","_number":2,"created":"2020-07-28 14:03:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/23/742523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/742523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/742523/2"}}},"commit":{"parents":[{"commit":"a25d505d7ca2c6188e5c1519e3af00360243eb64","subject":"Merge \"Fix a bug on list_unreleased_changes during tag retrieving\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a25d505d7ca2c6188e5c1519e3af00360243eb64"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-22 19:29:46.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-28 14:01:24.000000000","tz":-240},"subject":"Cinder: EOL Pike","message":"Cinder: EOL Pike\n\nAs announced on the mailing list [0,1,2], the Cinder project team\nwishes to transition the stable/pike branches of its deliverables\nfrom \u0027unmaintained\u0027 to EOL, and hereby requests that pike-eol tags\nbe applied to the stable/pike branches and the branches be deleted.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015838.html\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016050.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016076.html\n\nChange-Id: If266ab3efa5bdde3642d13136363573208b1e4d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/994d3f174914ff8747a3847aec6d579c2d2eb8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/994d3f174914ff8747a3847aec6d579c2d2eb8aa"}]},"branch":"refs/heads/master"},"fb6015c8e48a5f9489c4f46a893e386d88b1e71a":{"kind":"REWORK","_number":3,"created":"2020-07-29 11:29:13.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/23/742523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/742523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/742523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/742523/3"}}},"commit":{"parents":[{"commit":"b29d82d9a00367ffdf5b5001965ada525c03a9f0","subject":"Merge \"Victoria-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b29d82d9a00367ffdf5b5001965ada525c03a9f0"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-22 19:29:46.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-29 11:28:36.000000000","tz":-240},"subject":"Cinder: EOL Pike","message":"Cinder: EOL Pike\n\nAs announced on the mailing list [0,1,2], the Cinder project team\nwishes to transition the stable/pike branches of its deliverables\nfrom \u0027unmaintained\u0027 to EOL, and hereby requests that pike-eol tags\nbe applied to the stable/pike branches and the branches be deleted.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015838.html\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016050.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016076.html\n\nChange-Id: If266ab3efa5bdde3642d13136363573208b1e4d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb6015c8e48a5f9489c4f46a893e386d88b1e71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb6015c8e48a5f9489c4f46a893e386d88b1e71a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
