)]}'
{"id":"openstack%2Freleases~743420","triplet_id":"openstack%2Freleases~master~I1949deed094822d2c2c56659eadb1fc5ea6a59e5","project":"openstack/releases","branch":"master","topic":"safe_load","hashtags":[],"change_id":"I1949deed094822d2c2c56659eadb1fc5ea6a59e5","subject":"Switch to PyYaml safe_load","status":"MERGED","created":"2020-07-27 21:33:51.000000000","updated":"2020-08-03 14:01:17.000000000","submitted":"2020-08-03 14:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743420-1596463277891-3eeecb9b","meta_rev_id":"88ac1e6e79165d1e11d325d0f347440b498f81e0","_number":743420,"virtual_id_number":743420,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-03 13:32:01.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 14:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-03 08:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-07-29 09:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-03 08:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 09:31:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-03 08:44:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-03 14:01:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ff3be1ebc3d8590436284607a1df3524f24e165","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-27 21:33:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a84d9b1222f5f02b2ace51fc2c9c697caf525f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 21:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a16844a9126419ab74c124ec2258b4d : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e797da83b0ba4fdfafe0013a23f303db : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/271582044f2548499c7605c9f8b0ba4b : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9fbb21754c45a9beef53b79aea6f1f : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d9a30fbc35641e59df1cea92cded081 : SUCCESS in 4m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/770c22d6101742d4b860eb1caef193f0 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e97617c16d774fc79198520d2e50ee10 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a2c3392e236aff014f7eb433244a98d0ecab689b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 09:31:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nSide note, pyyaml is now safe by default [1]. Anyway these changes can\u0027t hurt.\n\n[1] https://github.com/yaml/pyyaml/pull/74/files","accounts_in_message":[],"_revision_number":1},{"id":"d66a2beb2c8b7a521136e29ff9b7f5c1e2d013b0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-03 08:44:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b55c26cfbac3271eb3308a882c835a1dbea887d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 08:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0073da01a57e5b0378052e31798b8c31ec86bc8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 09:12:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b75a16e4732492f805e563e8df4a3f6 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/546262088dbe44759f70a7d7c81017ab : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d62a4673586454ab6b3dc9124a1c515 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff0a6856b2343c6a3398ea101c102b7 : FAILURE in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/010143aec976487da94c5f5af6fe40d7 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d42de4d1e577453babfd2dc1022e21bf : FAILURE in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"58c1cfe2d4144381437f1e8022ebb1a15e28f95e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-03 11:51:17.000000000","message":"Patch Set 1:\n\nrecheck - after 0.7.0 release, hopefully not too soon","accounts_in_message":[],"_revision_number":1},{"id":"aeff6d44431d93eb01dc91b64a4c337bc463a5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:07:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60681fcd7d647a9b42d84d544fe6067 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/906e582b9116413ab215bea720e6b4a6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3faf3152772441f2abc144983cd08f47 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ea92e43f194ae6bdcba5e85df4a455 : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e860e14e7a045a887c626de85328b13 : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c5aaa38e41814209b302929a8d9ef087 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/042d88f22bac42f6bc18e36754b10a92 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d493b6abc106c89f0f9c7b8aec73904095e7a061","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 13:32:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4386900248e076e5c55d3505cc076c82b545ee85","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-03 13:39:01.000000000","message":"Patch Set 1:\n\nrecheck - now that 0.7.0 is required","accounts_in_message":[],"_revision_number":1},{"id":"72eaa3f9279267cc9eee808c9ea1a8c2b426be86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:47:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640bd8da7ac440e6b01c33d6aff11725 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5fb2f32cb98435ba9224dfc1c57463c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a883c58300824bce96a18cf24a3c5e8f : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7daa0ce9014e6ca3fca6b10b419807 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d0c88df326e4f169b770c919223701c : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6fd848b7ea304a1cbde1386a342ff8c8 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c97d640caa8742498c53c88f4bbd2c70 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fbe5abc6ed682dd26394639e865e0714a9837acc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:47:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7ffb7949132ffa2a1a4aabe60b5961cd6d2bd7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351ff73b2891450b8e00a602501b1714 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06090b28cacc4551b74346f6959c7b25 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/741301e8f0bd4f54b04451ff35115bac : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15e69a6655845beb32870d7c941c7c8 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc74ec63ae864e4eb685fe4304c75ee0 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9de0157920c442829fbd9ff7b0432235 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a8975497bb66a999698c0abdabe62b6df8997d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:01:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08cd1fb33c854be1b976a5f013cc9749352f085f","revisions":{"08cd1fb33c854be1b976a5f013cc9749352f085f":{"kind":"REWORK","_number":1,"created":"2020-07-27 21:33:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/743420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/743420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/743420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/743420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/743420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/743420/1"}}},"commit":{"parents":[{"commit":"b272ef96bacefb125dfb054d20ac477a7c448c24","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b272ef96bacefb125dfb054d20ac477a7c448c24"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-27 21:18:19.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-27 21:18:19.000000000","tz":-300},"subject":"Switch to PyYaml safe_load","message":"Switch to PyYaml safe_load\n\nThe load() call from PyYaml is considered a higher security risk in that\nit uses the FullLoader. safe_loade() is considered more safe by using\nthe SafeLoader instead.\n\nSince the 5.1 release of PyYaml added warning output when using load(),\nthis switches over to safe_load() to avoid the unnecessary noise.\n\nChange-Id: I1949deed094822d2c2c56659eadb1fc5ea6a59e5\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08cd1fb33c854be1b976a5f013cc9749352f085f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08cd1fb33c854be1b976a5f013cc9749352f085f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
