)]}'
{"id":"openstack%2Freleases~744473","triplet_id":"openstack%2Freleases~master~Ic2cb11ac59e35d4b41371092ca297e9e75282bfa","project":"openstack/releases","branch":"master","topic":"release-os-brick-3.2.1","hashtags":[],"change_id":"Ic2cb11ac59e35d4b41371092ca297e9e75282bfa","subject":"Release os-brick 3.2.1","status":"MERGED","created":"2020-08-03 13:40:32.000000000","updated":"2020-08-04 15:04:02.000000000","submitted":"2020-08-04 15:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744473-1596553442722-bf775c37","meta_rev_id":"a87d81b8050cec4b354bf200f541bfd13b6df1cb","_number":744473,"virtual_id_number":744473,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 15:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-04 14:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-04 14:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-03 13:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 14:51:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-04 15:04:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b07650ebf6f929ddefdc747c665d560c9a049ed4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-03 13:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3138c851dc30de7c1a27a2621b33b9cf0ae351f6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:41:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"18a604d4bdb457517719b75176ef57936a6883b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906d80490c6f4cddac187853e55535db : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0203b1d080194684a0206520db6a5b38 : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/330b5b43cf7e43cb923080508a23af09 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f63f223537f9072137ba8c5d03e39adfeb86a7f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-04 14:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8763c6276eb390a1ee574a7071d767e947bbd5c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 14:53:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"683bf2e21d4e05af4974b341ea23103feaba1ab4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:04:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18bc1cdf89384a1ef7e4154e810ed1ac9ef84a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:04:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87704d070b1e455abbe3b899fd9ab6e3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c233c421cd4f1ea590cd53f48d8439 : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3317f2266fcf427bbf68d7556f9d2b92 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d085cf9e6fa954264639bb0953982ab357d46ea","revisions":{"8d085cf9e6fa954264639bb0953982ab357d46ea":{"kind":"REWORK","_number":1,"created":"2020-08-03 13:40:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/73/744473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/744473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/744473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/744473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/744473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/744473/1"}}},"commit":{"parents":[{"commit":"4e944ea7b4d5456543f9da3286220f97a2366ec0","subject":"Require latest governance parsing code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e944ea7b4d5456543f9da3286220f97a2366ec0"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-03 13:39:48.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-03 13:39:48.000000000","tz":-240},"subject":"Release os-brick 3.2.1","message":"Release os-brick 3.2.1\n\nChange-Id: Ic2cb11ac59e35d4b41371092ca297e9e75282bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d085cf9e6fa954264639bb0953982ab357d46ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d085cf9e6fa954264639bb0953982ab357d46ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
