)]}'
{"id":"openstack%2Freleases~745329","triplet_id":"openstack%2Freleases~master~Id68796065bd287fb0b65a1b5bf62a9910689b813","project":"openstack/releases","branch":"master","topic":"sushy-tools-0120","hashtags":[],"change_id":"Id68796065bd287fb0b65a1b5bf62a9910689b813","subject":"Release sushy-tools 0.12.0 for Victoria","status":"MERGED","created":"2020-08-07 13:37:12.000000000","updated":"2020-08-10 12:19:43.000000000","submitted":"2020-08-10 12:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745329-1597061983963-6ccf36b0","meta_rev_id":"d918df6f1e8ad6df96e08754386628bf4f23fdcf","_number":745329,"virtual_id_number":745329,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 12:19:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-07 14:24:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-10 12:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-07 14:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-10 12:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-07 14:31:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 14:24:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-07 14:29:16.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-10 12:11:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-10 12:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ce51b6cbb5a14e9ec379d593d6558d41c0501a4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-07 13:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4593a5eac438c5688b3072e92083149f674b33b9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-07 13:47:03.000000000","message":"Patch Set 1: Code-Review-1\n\nThis release changes some user-visible behaviours, let\u0027s call it 0.12.0","accounts_in_message":[],"_revision_number":1},{"id":"8acadf2f0bbb9f860e6317f88fa8cdd333fb249b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902aa615056c4e2c83e99da1b741ac4e : SUCCESS in 10m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a09753e67a164ac1b8e09f29a9e6a273 : SUCCESS in 6m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/40cc2fcc4a71466998f4e32fbf7195bc : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"301e8a2eabd931dd8cc08d1b81fcd9826c6aeb13","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-07 14:04:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3cd403e0690bb8d7bde225362764be4f6820dbfb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-07 14:05:01.000000000","message":"Topic changed from sushy-tools-0112 to sushy-tools-0120","accounts_in_message":[],"_revision_number":2},{"id":"c453c14bfeaacbefa9857c7765339f985b170e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:18:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54878b86847491b906cfaf891623783 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13c2809994b24a8f842bfe9a297ce8d3 : SUCCESS in 4m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/786f9ed755c04ffc8464fa915b30ebe3 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"dbc7bba0a9d89f904e7f37fdd603cb9c7d7a2bf8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-07 14:24:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c9b42e09bcb48b62de065ca4e8f652afa7d03be7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-07 14:29:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e165326ba02d83dbbb2d2a89978030fd45351727","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:31:37.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f1c9b228fd389ec31c51df1028f5acdb033880f8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-10 12:11:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"169fc0c74d510a1e503348e5e5098bfe41237be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:12:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a099faca15728999a473d01d9cf6d4b6c468c2c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:19:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5357f3249d4d47c8bfd9def4026a8000 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f880cc986e4546811c85f435bb1d9f : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17e6dcc93be341f49f839280d17e54fb : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cfbf98a6253eeb6cd4cb86e3ed6ea0d508c62bc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:19:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ccffee6b6213490de8f91a60d7986d8ddcd3b43","revisions":{"669d43b327eff28daa69f5da8b4a9b24a7af4970":{"kind":"REWORK","_number":1,"created":"2020-08-07 13:37:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/29/745329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/745329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/745329/1"}}},"commit":{"parents":[{"commit":"bf9e35d5fff254456839c1aab250334de1c5df90","subject":"Release neutron-lib 2.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf9e35d5fff254456839c1aab250334de1c5df90"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-07 13:36:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-07 13:36:50.000000000","tz":120},"subject":"Release sushy-tools 0.11.2 for Victoria","message":"Release sushy-tools 0.11.2 for Victoria\n\nChange-Id: Id68796065bd287fb0b65a1b5bf62a9910689b813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/669d43b327eff28daa69f5da8b4a9b24a7af4970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/669d43b327eff28daa69f5da8b4a9b24a7af4970"}]},"branch":"refs/heads/master"},"3ccffee6b6213490de8f91a60d7986d8ddcd3b43":{"kind":"REWORK","_number":2,"created":"2020-08-07 14:04:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/29/745329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/745329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/745329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/745329/2"}}},"commit":{"parents":[{"commit":"bf9e35d5fff254456839c1aab250334de1c5df90","subject":"Release neutron-lib 2.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf9e35d5fff254456839c1aab250334de1c5df90"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-07 13:36:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-07 14:04:35.000000000","tz":120},"subject":"Release sushy-tools 0.12.0 for Victoria","message":"Release sushy-tools 0.12.0 for Victoria\n\nChange-Id: Id68796065bd287fb0b65a1b5bf62a9910689b813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ccffee6b6213490de8f91a60d7986d8ddcd3b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ccffee6b6213490de8f91a60d7986d8ddcd3b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
