)]}'
{"id":"openstack%2Freleases~745911","triplet_id":"openstack%2Freleases~master~I6ba030f678879d32f99680c68dcd4bff9cc00032","project":"openstack/releases","branch":"master","topic":"wallaby","hashtags":[],"change_id":"I6ba030f678879d32f99680c68dcd4bff9cc00032","subject":"Proposed release schedule for Wallaby (26w)","status":"MERGED","created":"2020-08-12 14:48:28.000000000","updated":"2020-08-27 16:19:16.000000000","submitted":"2020-08-27 16:19:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745911-1598545156926-210ac784","meta_rev_id":"76b8c169787a92b3e652214f7ae3db30c9532851","_number":745911,"virtual_id_number":745911,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 16:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 21:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-24 10:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-08-20 16:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2020-08-19 14:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-08-14 02:46:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-12 15:31:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2020-08-18 16:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-27 16:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 15:31:58.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2020-08-14 02:46:07.000000000","updated_by":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"reviewer":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"state":"REVIEWER"},{"updated":"2020-08-18 16:58:22.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-08-19 14:32:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-20 16:48:16.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-08-27 16:07:22.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-27 16:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"206029b82a1fd91393e9ac913397699e10ed8554","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-12 14:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d54d49edb58187d1ce344fb5bcd129d31d812379","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-12 14:51:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"53bd8c0fbc55170f038b856eb2cb669190c0b706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:16:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ea6602420aa4c3abe80d497f57b5777 : FAILURE in 15m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be238aaa4804d7f88997a158233e3fb : SUCCESS in 23m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6563a4e6fe5476c9b0387e751c6a626 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"257eef26e6a213ddf0b77b369ee3d6344ed70faf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-12 15:20:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a7db3f71a28a1ff940eb9e1a520c1cd93a86a40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:29:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a746f4692d547c7be7aeedbc1bcafa7 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e45e7d67874b3eb4086f64b39d475e : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca237c9357cc42acab8f7bcb0778f7de : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4bc54c816f22cb7d2ce21322fe6c999d1b8ed46d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-08-12 15:31:58.000000000","message":"Patch Set 2: Code-Review+1\n\nThis works really well downstream in Ubuntu, allowing inclusion of OpenStack Wallaby in Ubuntu 21.04.","accounts_in_message":[],"_revision_number":2},{"id":"73d642f6b98b528638b144f1bc9cfa500447592c","author":{"_account_id":20635,"name":"Ryan Beisner","email":"ryanbeisner@gmail.com","username":"ryan-beisner"},"date":"2020-08-14 02:46:07.000000000","message":"Patch Set 2: Code-Review+1\n\nThe cloud archive’s (UCA’s) inclusion of Wallaby aligns well with this.","accounts_in_message":[],"_revision_number":2},{"id":"8bf3a506ca047219e96c556d56aa2f17e3e6ae3a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-08-18 16:58:22.000000000","message":"Patch Set 2: Code-Review+1\n\nBased on the concerns raised on the longer schedule I think this makes sense.","accounts_in_message":[],"_revision_number":2},{"id":"567b7cf79140e7022c6001a04e00ae86e684e6b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 14:32:29.000000000","message":"Patch Set 2: Code-Review+2\n\nBoth schedules [1] LGTM but this one looks more consensual so +2 for this one.\n\n[1] https://review.opendev.org/#/c/744729/","accounts_in_message":[],"_revision_number":2},{"id":"4dd9fe79d5010741392626f0c261a4241c64d092","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-08-20 16:48:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ed8c7bd2861af3fd78b0de8028d93c10e56a7f15","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-24 10:05:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a048df4e34688905e841d523dbac94a66c4331c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 21:20:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"abc4a1ae3e150df02be7594d1ba33f53f617fa88","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-27 16:07:22.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68359ab80dc91876ce95e6995b65a61c24c221ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:09:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b949c0cb176aa144590f69722cdc4d2420dc924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:19:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4c2dd88d24a9bda5bc80d853b8bb9c1bd3984a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:19:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec567e978a14fffa6ca294801c3cf0b : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba51eac33c7246f4bb44582f823c0b4d : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dedbd7bdd0cc4758857e312577e1ff5c : SUCCESS in 7m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/846beb3c042e489582da942baf1273d5 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8060f963f6320e2fa524deb0fc189bb0e338271c","revisions":{"f4ac15563c7f1ebb33bb9367e2838de61167e4a5":{"kind":"REWORK","_number":1,"created":"2020-08-12 14:48:28.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/11/745911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/745911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/745911/1"}}},"commit":{"parents":[{"commit":"9e1067207c98a71ac459e009472b2f27adef7f43","subject":"Merge \"Release keystoneauth 4.2.1 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9e1067207c98a71ac459e009472b2f27adef7f43"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-04 14:37:05.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-12 14:47:12.000000000","tz":-300},"subject":"Proposed release schedule for Wallaby (26w)","message":"Proposed release schedule for Wallaby (26w)\n\nThis is a 26 week schedule for the Wallaby release. Planned release date\nis May 5, 2021. We don\u0027t have a date set for a first half 2021 event,\nbut it is expected to be some time shortly after this date.\n\nChange-Id: I6ba030f678879d32f99680c68dcd4bff9cc00032\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4ac15563c7f1ebb33bb9367e2838de61167e4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4ac15563c7f1ebb33bb9367e2838de61167e4a5"}]},"branch":"refs/heads/master"},"8060f963f6320e2fa524deb0fc189bb0e338271c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-12 14:51:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/11/745911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/745911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/745911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/745911/2"}}},"commit":{"parents":[{"commit":"9e1067207c98a71ac459e009472b2f27adef7f43","subject":"Merge \"Release keystoneauth 4.2.1 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9e1067207c98a71ac459e009472b2f27adef7f43"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-04 14:37:05.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-12 14:51:08.000000000","tz":-300},"subject":"Proposed release schedule for Wallaby (26w)","message":"Proposed release schedule for Wallaby (26w)\n\nThis is a 26 week schedule for the Wallaby release. Planned release date\nis April 14, 2021. We don\u0027t have a date set for a first half 2021 event,\nbut it is expected to be some time shortly after this date.\n\nChange-Id: I6ba030f678879d32f99680c68dcd4bff9cc00032\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8060f963f6320e2fa524deb0fc189bb0e338271c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8060f963f6320e2fa524deb0fc189bb0e338271c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
