)]}'
{"id":"openstack%2Freleases~746187","triplet_id":"openstack%2Freleases~master~Idea391efe6031ce90c0af386b63979ff27845988","project":"openstack/releases","branch":"master","topic":"victoria-cwi","hashtags":[],"change_id":"Idea391efe6031ce90c0af386b63979ff27845988","subject":"Switch Adjutant to cycle-with-rc","status":"MERGED","created":"2020-08-13 16:18:01.000000000","updated":"2020-08-18 13:27:30.000000000","submitted":"2020-08-18 13:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746187-1597757250033-7628fdb8","meta_rev_id":"4ba0f435e9ac01e0e0bf174d34afd094b6ae6ddb","_number":746187,"virtual_id_number":746187,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 13:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-18 13:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-17 23:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-18 13:19:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-17 23:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 23:09:42.000000000","updated_by":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"reviewer":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"state":"REVIEWER"},{"updated":"2020-08-18 13:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26d8090b2dccbc42c2a1b6fd5983358c61773b45","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-13 16:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4896002d5dc2f7eac4bff6f961864949745cc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:35:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4705980b7c4d019cfe5e0b8139ed8d : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b643a15db50a461a8d92b3d2e691cc6f : FAILURE in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/09d3eff383814f0687f5a1d8f50db18f : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bfe6a53dd6394bcf915ac05958d03cf381ffccd9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-13 16:36:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9be265c74fb782e8840ee0f9f14d0bb7fdcc77dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:48:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61237297138444284fee8f17495b799 : SUCCESS in 9m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e31c519f8984447f92fd3054599be244 : SUCCESS in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/be280482740a4117bcab54d326a38f22 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b674257b0970078cb7f03292a7532e02ac0145ee","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-14 03:17:12.000000000","message":"Patch Set 2: Code-Review-1\n\nI don\u0027t particularly want to switch back to cycle with RC because that implies mid cycle tags as well, which I can\u0027t always commit to.\n\nI will though be tagging a new release before RC1 (roughly Sept 22ish).\n\nSadly between internal work and covid, Adjutant has mostly been ignored upstream unless a bug was found. So there isn\u0027t much to tag since Ussuri anyway feature wise bar a few minor changes one of which is still waiting on review.\n\nPeople are using it, but no external developers contributing.\n\nTechnically I can tag now and just include my config library changes, but I\u0027d ideally like to include: https://review.opendev.org/#/c/725746/","accounts_in_message":[],"_revision_number":2},{"id":"fa45215c53d00ceb780e61412cac343c3ba9b878","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 12:47:27.000000000","message":"Patch Set 2:\n\n\u003e I don\u0027t particularly want to switch back to cycle with RC because\n \u003e that implies mid cycle tags as well, which I can\u0027t always commit\n \u003e to.\n \u003e \nWe can definitely abandon this, but first I wanted to make sure there was no confusion.\n\nWith cycle-with-rc, there is no requirement to do releases during the cycle. Just a release candidate at the RC1 deadline, which can then be retagged for the final release if there are no other changes. So I think ultimately, it ends up being less work overall for you.\n\nWill wait to abandon to see if that makes any difference.","accounts_in_message":[],"_revision_number":2},{"id":"bc1ae4703b92912b3135759cbb0eade665468348","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-16 23:43:44.000000000","message":"Patch Set 2:\n\n\u003e \u003e I don\u0027t particularly want to switch back to cycle with RC because\n \u003e \u003e that implies mid cycle tags as well, which I can\u0027t always commit\n \u003e \u003e to.\n \u003e \u003e\n \u003e We can definitely abandon this, but first I wanted to make sure\n \u003e there was no confusion.\n \u003e \n \u003e With cycle-with-rc, there is no requirement to do releases during\n \u003e the cycle. Just a release candidate at the RC1 deadline, which can\n \u003e then be retagged for the final release if there are no other\n \u003e changes. So I think ultimately, it ends up being less work overall\n \u003e for you.\n \u003e \n \u003e Will wait to abandon to see if that makes any difference.\n\nI swear there used to be some midpoint test releases we had to tag which I missed once a while back, and if we didn\u0027t meet those we can\u0027t be included in the release even if we tag an RC1?","accounts_in_message":[],"_revision_number":2},{"id":"764160bc8e3549b2aba2b9624b9d41fcf690f941","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 12:12:42.000000000","message":"Patch Set 2:\n\n\u003e I swear there used to be some midpoint test releases we had to tag\n \u003e which I missed once a while back, and if we didn\u0027t meet those we\n \u003e can\u0027t be included in the release even if we tag an RC1?\n\nA few release cycles ago we dropped the need to do beta releases at each milestone. It is now just the release candidate at the RC1 deadline, then the final release.","accounts_in_message":[],"_revision_number":2},{"id":"735f7bcd722bcb554dd06e98fb17f85d4799d83d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 23:09:42.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e \u003e I swear there used to be some midpoint test releases we had to\n \u003e tag\n \u003e \u003e which I missed once a while back, and if we didn\u0027t meet those we\n \u003e \u003e can\u0027t be included in the release even if we tag an RC1?\n \u003e \n \u003e A few release cycles ago we dropped the need to do beta releases at\n \u003e each milestone. It is now just the release candidate at the RC1\n \u003e deadline, then the final release.\n\nThen cycle-with-rc would work. I think there are some cases where tagging outside of that might be useful, but I can tag beta releases if I do need to, or people can be silly and do what we do, which is run master...","accounts_in_message":[],"_revision_number":2},{"id":"35a1bf6544a43e4f586c71f3ce35ca5f6a1caad0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 23:10:49.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"2e225b26ce252d0d4bc1d2257898500f2496e5f7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 13:19:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Adrian, this does seem a better fit for now. And you are right, beta releases can always be done if needed. But no need to do them just for the sake of doing them.","accounts_in_message":[],"_revision_number":2},{"id":"b2946ebd3ee01224aad9e650e74d467c2c46c69d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:19:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14cf1ba2f0f863bf9470460d0eba3a006e9add46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:27:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00df43cdd364415d80a11650d85b7c70 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e8984230b54ef7a34031525a62761d : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ddcc37020bba479293183bfcec42abb6 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"38b62151708c8bcf836f24b4f9bd110b1366b04d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a80a4d898b8079befa8e56e3f39038cdcbbdd47a","revisions":{"10d6620c544cf499c20367900170ad17a9a3256e":{"kind":"REWORK","_number":1,"created":"2020-08-13 16:18:01.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/87/746187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/746187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/746187/1"}}},"commit":{"parents":[{"commit":"b52cec85040e11f62931f6f9be5638c35aeceb28","subject":"Merge \"release stable Glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b52cec85040e11f62931f6f9be5638c35aeceb28"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-13 16:17:43.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-13 16:17:43.000000000","tz":-300},"subject":"Switch Adjutant to cycle-with-rc","message":"Switch Adjutant to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nAdjutant only released once last cycle, and has not done a\nrelease yet this cycle, this proposes changing the release model to\ncycle-with-rc so the expectation is only that it will release an RC by\nthe end of the cycle prior to doing the final release.\n\nThis is just a suggestion for the team to consider. The important point\nis past the milestone-2 point, downstream planning starts happening\nfor what will be included in the release. We just want to try to avoid\nthe situation where downstream expects something to be included, but we\nneed to unexpectedly drop it from the coordinated release late in the\ncycle.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: Idea391efe6031ce90c0af386b63979ff27845988\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10d6620c544cf499c20367900170ad17a9a3256e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10d6620c544cf499c20367900170ad17a9a3256e"}]},"branch":"refs/heads/master"},"a80a4d898b8079befa8e56e3f39038cdcbbdd47a":{"kind":"REWORK","_number":2,"created":"2020-08-13 16:36:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/87/746187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/746187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/746187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/746187/2"}}},"commit":{"parents":[{"commit":"b52cec85040e11f62931f6f9be5638c35aeceb28","subject":"Merge \"release stable Glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b52cec85040e11f62931f6f9be5638c35aeceb28"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-13 16:17:43.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-13 16:36:46.000000000","tz":-300},"subject":"Switch Adjutant to cycle-with-rc","message":"Switch Adjutant to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nAdjutant only released once last cycle, and has not done a\nrelease yet this cycle, this proposes changing the release model to\ncycle-with-rc so the expectation is only that it will release an RC by\nthe end of the cycle prior to doing the final release.\n\nThis is just a suggestion for the team to consider. The important point\nis past the milestone-2 point, downstream planning starts happening\nfor what will be included in the release. We just want to try to avoid\nthe situation where downstream expects something to be included, but we\nneed to unexpectedly drop it from the coordinated release late in the\ncycle.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: Idea391efe6031ce90c0af386b63979ff27845988\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a80a4d898b8079befa8e56e3f39038cdcbbdd47a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a80a4d898b8079befa8e56e3f39038cdcbbdd47a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
