)]}'
{"id":"openstack%2Freleases~746246","triplet_id":"openstack%2Freleases~master~Ia400cbbcf42d72c3e68d5ecf19e5cb8a4181bdd7","project":"openstack/releases","branch":"master","topic":"release_osa","hashtags":[],"change_id":"Ia400cbbcf42d72c3e68d5ecf19e5cb8a4181bdd7","subject":"Release OpenStack-Ansible Ussuri RC3","status":"MERGED","created":"2020-08-14 00:11:57.000000000","updated":"2020-08-14 15:14:52.000000000","submitted":"2020-08-14 15:14:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746246-1597418092607-ff0595dd","meta_rev_id":"c76cac3e5700ec8846e58489c0ad50d2fbc3aca0","_number":746246,"virtual_id_number":746246,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 15:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-14 15:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-14 15:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-14 00:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 15:04:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-14 15:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fbaecdb558262aeb3e9e7bd4c1714661ca1f7d65","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 00:11:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"392b02976af12ee5b8d0933174ffbd48444fc5d7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 00:12:51.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"31e3b77ffde58904fea2e1547b4cb63b5db93be5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 00:14:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"37b0ca18b60b39e3f6863045c211445bbe819ccd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 00:15:15.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"287cfd16febb5590e3f9074c8a1a09932421a3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 00:22:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f2bc0907904e3e9744a73d8afb8dcc : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/293d4a1cce5c406d8280d3a4baab74ba : SUCCESS in 4m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/40612832cfd94f14a8b9f8e7e28fa28d : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f151d6acb88f9a5e59f2fdd56d90db18ac517b2f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 11:55:22.000000000","message":"Patch Set 3:\n\nAbandon this?","accounts_in_message":[],"_revision_number":3},{"id":"2a06dd64388cb3c11a5e1c8082df8c856219e4f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 15:04:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3af76b92b7fee57d8889cb5e12c9b3194b3f9937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:04:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b511a7cb475e3694824ad104e49f7f8a907809e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:14:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"62e82799c431c2dd9c4a6b89abab63cfb2cead81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:14:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b981314827449d881283eacfc627780 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c6b791456e4c0bbd471ca1f3a8aea9 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29be28eb8d6a467aa58c7b3b3accd808 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c9d633ad170e8b757363cef5f26c0d1dac20ede","revisions":{"efcb17eb9e3e275f77d264ff9287546404598be0":{"kind":"REWORK","_number":1,"created":"2020-08-14 00:11:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/746246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/746246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/746246/1"}}},"commit":{"parents":[{"commit":"b520f234cd659d25b2b9c8c28be1318fcd8791ef","subject":"Merge \"Release ironic-prometheus-exporter 2.1.0 for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b520f234cd659d25b2b9c8c28be1318fcd8791ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-13 22:06:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-13 22:13:16.000000000","tz":180},"subject":"Release OpenStack-Ansible Ussuri RC3","message":"Release OpenStack-Ansible Ussuri RC3\n\nThis will be the final RC for OpenStack-\nAnsible.\n\nChange-Id: Ia400cbbcf42d72c3e68d5ecf19e5cb8a4181bdd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efcb17eb9e3e275f77d264ff9287546404598be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efcb17eb9e3e275f77d264ff9287546404598be0"}]},"branch":"refs/heads/master"},"f35daf497689083432204a0774b136c08956ab9d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-14 00:14:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/746246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/746246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/746246/2"}}},"commit":{"parents":[{"commit":"b520f234cd659d25b2b9c8c28be1318fcd8791ef","subject":"Merge \"Release ironic-prometheus-exporter 2.1.0 for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b520f234cd659d25b2b9c8c28be1318fcd8791ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-13 22:06:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-13 22:06:30.000000000","tz":180},"subject":"Release OpenStack-Ansible Ussuri RC3 This will be the final RC for OpenStack- Ansible.","message":"Release OpenStack-Ansible Ussuri RC3\nThis will be the final RC for OpenStack-\nAnsible.\n\nChange-Id: Ia400cbbcf42d72c3e68d5ecf19e5cb8a4181bdd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f35daf497689083432204a0774b136c08956ab9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f35daf497689083432204a0774b136c08956ab9d"}]},"branch":"refs/heads/master"},"7c9d633ad170e8b757363cef5f26c0d1dac20ede":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-14 00:15:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/746246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/746246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/746246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/746246/3"}}},"commit":{"parents":[{"commit":"b520f234cd659d25b2b9c8c28be1318fcd8791ef","subject":"Merge \"Release ironic-prometheus-exporter 2.1.0 for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b520f234cd659d25b2b9c8c28be1318fcd8791ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-13 22:06:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-14 00:15:11.000000000","tz":0},"subject":"Release OpenStack-Ansible Ussuri RC3","message":"Release OpenStack-Ansible Ussuri RC3\n\nThis will be the final RC for OpenStack-\nAnsible.\n\nChange-Id: Ia400cbbcf42d72c3e68d5ecf19e5cb8a4181bdd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c9d633ad170e8b757363cef5f26c0d1dac20ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c9d633ad170e8b757363cef5f26c0d1dac20ede"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
