)]}'
{"id":"openstack%2Freleases~746345","triplet_id":"openstack%2Freleases~master~If3fa129a93814705fa98d58d860d1dfc18393e73","project":"openstack/releases","branch":"master","topic":"cycle_highlights","hashtags":[],"change_id":"If3fa129a93814705fa98d58d860d1dfc18393e73","subject":"Add Cycle Highlights to Victoria","status":"MERGED","created":"2020-08-14 19:18:50.000000000","updated":"2020-08-18 13:30:51.000000000","submitted":"2020-08-18 13:30:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746345-1597757451342-ea02c699","meta_rev_id":"7502fe31bf8e506d2746aced70036a5f3c0c86d7","_number":746345,"virtual_id_number":746345,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 13:30:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-18 13:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-18 13:19:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 13:19:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-18 13:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"730be37a251ab0daf6cdf8eae81b28c0cea92e84","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-08-14 19:18:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b59a0ef31fd2fd276c1a8c37c1ec90d572b5d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 19:23:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ba9888bb8e433b9e5da2ee066dcf0e : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"39a478e7e2defb4a2ed051d965c7b0a11c6c1c73","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 19:45:11.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nrecheck\n\nPypi issue maybe? Docs build fine for me locally with a fresh venv.","accounts_in_message":[],"_revision_number":1},{"id":"b1561a751f090a11add5162e81fc863e112139b1","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-08-14 19:59:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c742a69182a8e08fa649abe48678ae6569c5b8f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-08-17 16:00:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b307d6d312bd7ac52fee7dc5c79715edbb9d4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:18:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4800a20650444291b5fcb9a2985956 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cc4be120babd3b6f24204c5745344e1b65bd27f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 13:19:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba308b52e69eba41b19f0ac9abf8ff4cba2bb4da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:20:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4807d24ec9ff885c5a9e326043e2db53de9004fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:30:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0c9ab223c4e9e0532b5adb190c2c3767a555674e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 13:30:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc0a4f145504cae83c1ff23060d7ad3 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51652adb0f8e447b8eb54c6f5d9a1f55 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6018e3b095fc995d9e3bfbe719db9c26f21a423","revisions":{"f2b92acaef49a76f2037f86738c50a2dee16d2fe":{"kind":"REWORK","_number":1,"created":"2020-08-14 19:18:50.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/45/746345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/746345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/746345/1"}}},"commit":{"parents":[{"commit":"2f5e4ca9c27db745a6a3a6ad7c4deab20c22b3a8","subject":"Release OpenStack-Ansible Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f5e4ca9c27db745a6a3a6ad7c4deab20c22b3a8"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-08-14 19:18:16.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-08-14 19:18:49.000000000","tz":-420},"subject":"Add Cycle Highlights to Victoria","message":"Add Cycle Highlights to Victoria\n\nChange-Id: If3fa129a93814705fa98d58d860d1dfc18393e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2b92acaef49a76f2037f86738c50a2dee16d2fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2b92acaef49a76f2037f86738c50a2dee16d2fe"}]},"branch":"refs/heads/master"},"e6018e3b095fc995d9e3bfbe719db9c26f21a423":{"kind":"REWORK","_number":2,"created":"2020-08-17 16:00:46.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/45/746345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/746345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/746345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/746345/2"}}},"commit":{"parents":[{"commit":"2f5e4ca9c27db745a6a3a6ad7c4deab20c22b3a8","subject":"Release OpenStack-Ansible Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f5e4ca9c27db745a6a3a6ad7c4deab20c22b3a8"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-08-14 19:18:16.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-08-17 16:00:35.000000000","tz":-420},"subject":"Add Cycle Highlights to Victoria","message":"Add Cycle Highlights to Victoria\n\nChange-Id: If3fa129a93814705fa98d58d860d1dfc18393e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6018e3b095fc995d9e3bfbe719db9c26f21a423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6018e3b095fc995d9e3bfbe719db9c26f21a423"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
