)]}'
{"id":"openstack%2Freleases~747186","triplet_id":"openstack%2Freleases~master~Ie0791724cff7308cb419ba046e65e0f48f70b67e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie0791724cff7308cb419ba046e65e0f48f70b67e","subject":"Source the related tox environment","status":"ABANDONED","created":"2020-08-20 14:39:06.000000000","updated":"2020-08-20 20:39:30.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93592395c05d9715247ad2e45d25306c84b43b07","_number":747186,"virtual_id_number":747186,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-20 18:40:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-08-20 20:39:14.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-20 18:50:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 18:40:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-20 18:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cca4e11ca81ed0bb1f2b3c302f193a13165ec38e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 14:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6d56b65f3da345d3fb060fdd9beba1788dd452f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-20 14:41:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nIs there a reason we need to support this?","accounts_in_message":[],"_revision_number":1},{"id":"94e94bdcfde713920a9d882c41b359705937a9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:35:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fde0ff9a7de4422ea5430471a2769927 : RETRY_LIMIT in 1m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba012a89177148738396fd34fbdb33d3 : RETRY_LIMIT in 1m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d654248ee6444c0eb8084e1918ba7136 : RETRY_LIMIT in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f9879dfc9a414a9741f42a2d499e10 : RETRY_LIMIT in 1m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67893639589f4c9ba7a441e4eee9f7d6 : RETRY_LIMIT in 1m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa758cd254a747ce8bc105cd17a4426b : RETRY_LIMIT in 2m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9749b59c1c24e3899739b2cecedf5f9 : RETRY_LIMIT in 2m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cfc1d297ce29414bbee9481fd2ce6e9a : RETRY_LIMIT in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6fb9479d54ffc04d6a337f1c0d4e72d3a83f7b3a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 18:36:12.000000000","message":"Patch Set 1:\n\nrecheck\n\n \u003e (1 comment)\n \u003e \n \u003e Is there a reason we need to support this?\n\nThe main reason before these is that I personally often call this command from other repository and I noticed that some conflicts can occure when this other repo also use tox, so to avoid conflicts I prefer to use full path. I don\u0027t think that I\u0027m the one to call this kind of command from an external directory, and I don\u0027t it can hurt, we just ensure to source the right venv to properly retrieve our python comands (e.g list-unreleased-changes) even outside of this repo.","accounts_in_message":[],"_revision_number":1},{"id":"c8c8e090291efe24863b51e5444f230d5f2f9542","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 18:36:37.000000000","message":"Patch Set 1:\n\n\u003e recheck\n \u003e \n \u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e Is there a reason we need to support this?\n \u003e \n \u003e The main reason before these is that I personally often call this\n \u003e command from other repository and I noticed that some conflicts can\n \u003e occure when this other repo also use tox, so to avoid conflicts I\n \u003e prefer to use full path. I don\u0027t think that I\u0027m the one to call\n \u003e this kind of command from an external directory, and I don\u0027t it can\n \u003e hurt, we just ensure to source the right venv to properly retrieve\n \u003e our python comands (e.g list-unreleased-changes) even outside of\n \u003e this repo.\n\ns/before/behind/","accounts_in_message":[],"_revision_number":1},{"id":"2cbac645a42db52b326389b48bc826fff8e52c61","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 18:37:56.000000000","message":"Patch Set 1:\n\n\u003e \u003e recheck\n \u003e \u003e\n \u003e \u003e \u003e (1 comment)\n \u003e \u003e \u003e\n \u003e \u003e \u003e Is there a reason we need to support this?\n \u003e \u003e\n \u003e \u003e The main reason before these is that I personally often call this\n \u003e \u003e command from other repository and I noticed that some conflicts\n \u003e can\n \u003e \u003e occure when this other repo also use tox, so to avoid conflicts I\n \u003e \u003e prefer to use full path. I don\u0027t think that I\u0027m the one to call\n \u003e \u003e this kind of command from an external directory, and I don\u0027t it\n \u003e can\n \u003e \u003e hurt, we just ensure to source the right venv to properly\n \u003e retrieve\n \u003e \u003e our python comands (e.g list-unreleased-changes) even outside of\n \u003e \u003e this repo.\n \u003e \n \u003e s/before/behind/\nAaargghhh sorry... \n\nReword:\n\nThe main reason behind these changes is that I personally often call this command from other repository and I noticed that some conflicts can occure when this other repo also use tox, so to avoid conflicts I prefer to use full path. I don\u0027t think that I\u0027m the one to call this kind of command from an external directory, and I don\u0027t it can hurt, we just ensure to source the right venv to properly retrieve our python comands (e.g list-unreleased-changes) even outside of this repo.","accounts_in_message":[],"_revision_number":1},{"id":"8bf798e76f592a5599ceb96acf02b11450fe9202","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-20 18:40:58.000000000","message":"Patch Set 1:\n\nWell, it\u0027s been that way for years, so you may actually be the only one.\n\nBut don\u0027t recheck until you\u0027ve figured out how to address the issue pointed out in the change.","accounts_in_message":[],"_revision_number":1},{"id":"b0f5345dd02638e5725c56a20df9f67576e677f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:50:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c86bde22b8c4bae9267b0350537eaad : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79453f03f2df4570af4987e851e3c9e0 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e20e98721d7a4759b4917d4ddba05684 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd26070c4874b71824dec94adcf5182 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d4ce22562ab4302a3f68313679a17e0 : SUCCESS in 4m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cabb4963f9e0426dabdc56e15f6455ed : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c10b40716c4307ba07c48e30e07c89 : SUCCESS in 5m 48s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6b726f8d95b4477eb498ab28e69a0241 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a7a4928573521d89c0b6b31b302d0165750f9828","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 20:39:14.000000000","message":"Patch Set 1:\n\n\u003e Well, it\u0027s been that way for years, so you may actually be the only\n \u003e one.\n\nThanks for your feedback, I think I can easily survive without these changes, so let\u0027s drop them, sorry for the wasted time.\n\n \u003e But don\u0027t recheck until you\u0027ve figured out how to address the issue\n \u003e pointed out in the change.","accounts_in_message":[],"_revision_number":1},{"id":"93592395c05d9715247ad2e45d25306c84b43b07","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 20:39:30.000000000","message":"Abandoned\n\nNot really needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7af5a017a3c53932a5abb09c08475a1f6ca464a","revisions":{"c7af5a017a3c53932a5abb09c08475a1f6ca464a":{"kind":"REWORK","_number":1,"created":"2020-08-20 14:39:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/747186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/747186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/747186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/747186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/747186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/747186/1"}}},"commit":{"parents":[{"commit":"6c9d1d79f133a0852bd842c45e2a1262a4df4033","subject":"Merge \"Release Ironic 13.0.6 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c9d1d79f133a0852bd842c45e2a1262a4df4033"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-20 14:32:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-20 14:38:03.000000000","tz":120},"subject":"Source the related tox environment","message":"Source the related tox environment\n\nThese changes allow us to call script from anywhere even if we are\nin a different repo that also contains a tox virtual environment.\n\nWith these changes we ensure us to load the right env by retrieving\nthe full path relative to our script file.\n\nChange-Id: Ie0791724cff7308cb419ba046e65e0f48f70b67e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7af5a017a3c53932a5abb09c08475a1f6ca464a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7af5a017a3c53932a5abb09c08475a1f6ca464a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
