)]}'
{"id":"openstack%2Freleases~747219","triplet_id":"openstack%2Freleases~master~Ia8ad4e6851d2831a86fd681a31ed5b4131aad26f","project":"openstack/releases","branch":"master","topic":"neutron-15.1.1","hashtags":[],"change_id":"Ia8ad4e6851d2831a86fd681a31ed5b4131aad26f","subject":"Release Neutron 15.1.1 for Train","status":"ABANDONED","created":"2020-08-20 15:38:22.000000000","updated":"2020-08-26 08:48:12.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8efc9b5fbc925b6a00319247be4c14fa993c7e92","_number":747219,"virtual_id_number":747219,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-26 08:40:55.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-20 17:50:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":-1,"date":"2020-08-21 10:09:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 17:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 10:09:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-26 08:40:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"608398fc50d6490b50f5bea73b3b4eea587d7140","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-20 15:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"193594022edd743cfecc90eb7ac57b55f4a5c038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 17:50:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf71f5bb5764108b1ec421cb873787c : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82f18b68d68d4601b1c9d24e46a36c54 : SUCCESS in 17m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7268d69300749859fa2d94bfe82cd42 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c7d69acc858b296577e1aec9414d4fafdc508195","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 10:09:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nHello,\n\n-1 to highlight the minore version needs if we decide to continue with that (see my inline comment).","accounts_in_message":[],"_revision_number":1},{"id":"373e1850112080309b3a2cc8367e1a4aa3824029","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:38:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd8d249e1765ce58f4f718ae988f58e9af474379","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:40:55.000000000","message":"Patch Set 1:\n\nCatching up on gerrit backlog, there is in fact https://review.opendev.org/#/c/747869 now for 15.2.0 by slaweq","accounts_in_message":[],"_revision_number":1},{"id":"ab6fd2a29d95599a42983e8be5d48db9a03ded9a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-26 08:48:12.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/#/c/747869","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65b6a68651f46a48e3398a8e10980edb9f6e26c3","revisions":{"65b6a68651f46a48e3398a8e10980edb9f6e26c3":{"kind":"REWORK","_number":1,"created":"2020-08-20 15:38:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/19/747219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/747219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/747219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/747219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/747219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/747219/1"}}},"commit":{"parents":[{"commit":"6c9d1d79f133a0852bd842c45e2a1262a4df4033","subject":"Merge \"Release Ironic 13.0.6 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c9d1d79f133a0852bd842c45e2a1262a4df4033"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-08-20 15:36:08.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-08-20 15:36:08.000000000","tz":120},"subject":"Release Neutron 15.1.1 for Train","message":"Release Neutron 15.1.1 for Train\n\nThere is a new patch merged [1] in Neutron that is required by\nnetworking-ovn unittests to pass. The reason patch in networking-ovn was\nmerged was because of Depend-On reference, thus the ovn patch had what\nit needs but now it installs Neutron 15.1.0 that does not contain [1].\n\n[1] https://opendev.org/openstack/neutron/commit/8f1912a1d195dc69eed9d9aa8ff8c4ef12448b75\n\nChange-Id: Ia8ad4e6851d2831a86fd681a31ed5b4131aad26f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65b6a68651f46a48e3398a8e10980edb9f6e26c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65b6a68651f46a48e3398a8e10980edb9f6e26c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
