)]}'
{"id":"openstack%2Freleases~747244","triplet_id":"openstack%2Freleases~master~I9c538ff9996ca95b908b733326ed6872049b250c","project":"openstack/releases","branch":"master","topic":"oslo-independent-2020-08-20","hashtags":[],"change_id":"I9c538ff9996ca95b908b733326ed6872049b250c","subject":"Release pbr 5.5.0","status":"ABANDONED","created":"2020-08-20 16:27:32.000000000","updated":"2020-08-21 08:43:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0f39bf14cf9494b782b88079a9540a2bf4b3617","_number":747244,"virtual_id_number":747244,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-20 18:03:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-20 16:42:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 18:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edc28235621c74a7876d1b638830eb74aa05153c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-20 16:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f248668842563107d2b31e826251faff943e752","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:42:39.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"31b3fb602f199f89488e7edc4d0bae421783b353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:03:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e35ea3ebf504b63ac5a638e1df33f57 : SUCCESS in 10m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cc80750118a4f708a2af1d0ce9ebadf : FAILURE in 10m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/92ffb74eda6f4e569064e744820c6ec6 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":1},{"id":"41a16b9b8a22994ecee8be3cab0297524a2039d9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 08:43:36.000000000","message":"Abandoned\n\nBy looking to the embed changes more deeply I think we can continue to wait a bit before release a new version PBR.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6abf247652d4451a5845ea98c685a5a64c7e77d","revisions":{"c6abf247652d4451a5845ea98c685a5a64c7e77d":{"kind":"REWORK","_number":1,"created":"2020-08-20 16:27:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/747244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/747244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/747244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/747244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/747244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/747244/1"}}},"commit":{"parents":[{"commit":"6c9d1d79f133a0852bd842c45e2a1262a4df4033","subject":"Merge \"Release Ironic 13.0.6 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c9d1d79f133a0852bd842c45e2a1262a4df4033"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-20 16:02:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-20 16:02:20.000000000","tz":120},"subject":"Release pbr 5.5.0","message":"Release pbr 5.5.0\n\nMany patches have been merged recently it could be worth to make a new\nrelease for them.\n\nChange-Id: I9c538ff9996ca95b908b733326ed6872049b250c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6abf247652d4451a5845ea98c685a5a64c7e77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6abf247652d4451a5845ea98c685a5a64c7e77d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
