)]}'
{"id":"openstack%2Freleases~747873","triplet_id":"openstack%2Freleases~master~Iffb08675b6911336acc5e2a7dee3665282976441","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iffb08675b6911336acc5e2a7dee3665282976441","subject":"Release neutron 14.3.1 for stable/stein","status":"MERGED","created":"2020-08-25 09:12:03.000000000","updated":"2020-08-27 09:05:14.000000000","submitted":"2020-08-27 09:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747873-1598519114253-f4e8b138","meta_rev_id":"8ce8dd803231db5eae0cbfc3b2a9ad17467aecf8","_number":747873,"virtual_id_number":747873,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-08-27 08:14:09.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 09:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-27 08:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-08-27 08:13:37.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-08-27 08:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-08-27 08:21:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-27 08:16:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 08:38:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-26 20:50:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 08:17:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-27 08:21:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-27 09:05:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"923cd220ed79192ae90de475f0d0fc8ca896321e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-25 09:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b52a4c6fa3fe679c2d7c00ae4740eb25ee480d0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:13:29.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cf3480ff374dbf146ea1025ec60b86ecae341435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:40:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72b336739c4458a9f571202d8fc0026 : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dd0ebb212ad4620a24539682b39f91a : SUCCESS in 21m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d4a2da2e86e4b678047ab28808227a5 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cafab29d7a6fa6adb49da806b91528c77940ece3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:25:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f0cd4fdd9aa33a20d301ae245015a8ca1c096794","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:38:41.000000000","message":"Patch Set 1: -Code-Review\n\nSee comment in train realease on minor bump:\nhttps://review.opendev.org/#/c/747219","accounts_in_message":[],"_revision_number":1},{"id":"2cdbd53bb07e851f947dea78b193066f77bd3e8b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-26 08:48:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9de866a3c36979e64c83df828ac6affdfe8bff6b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-26 08:50:13.000000000","message":"Patch Set 2:\n\nBumped minor version due to https://review.opendev.org/#/c/740216/","accounts_in_message":[],"_revision_number":2},{"id":"935e99bc5c9f5a48127c87138c639a8670df7dcd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-26 08:50:34.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d554809ec8e880aced0176036142c546959fc4c6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:50:37.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"e9cc0328b05d959c120a34e61750b5e1229301aa","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:51:47.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"52e56f6e168aab1d4034fe152f291c53ddaefa2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:03:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3735ee0480364be98db523cfbe48df7a : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e270349f089473083924cbf5544b114 : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/95a2468277cb41c285925b96c07b8d49 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"18baa917e152013eba92c49ab055f8764f4f48e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 09:03:08.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nHello\n\n \u003e Bumped minor version due to https://review.opendev.org/#/c/740216/\n\nI don\u0027t seen these changes embed here, maybe I missed something, here is the changes that will be released:\n\n```\n$ git log --oneline 14.3.0..c65d1329e16ae3677c975f983e3920f50ed90a02\nc65d1329e1 (HEAD -\u003e stein, origin/stable/stein) Merge \"port_forwarding: validate args before invoking db update\" into stable/stein\nb3647782e4 port_forwarding: validate args before invoking db update\nfda4a375c4 [Stable only] Drop *-master jobs\n4ceee76477 Merge \"Fix Traceback when running neutron-ipset-cleanup tool\" into stable/stein\nb290c53716 Merge \"[Security] fix allowed-address-pair 0.0.0.0/0 issue\" into stable/stein\nb434f339e3 Merge \"Don\u0027t check if any bridges were recrected when OVS was restarted\" into stable/stein\nc2311465bf Fix validation of IPv6 subnets with external RAs\na93e6c922c Don\u0027t check if any bridges were recrected when OVS was restarted\n18a171808b [Security] fix allowed-address-pair 0.0.0.0/0 issue\n0d9ec2b758 Fix Traceback when running neutron-ipset-cleanup tool\n8181c5dbfe Ensure drop flows on br-int at agent startup for DVR too\n65a6010da7 Merge \"Optionally use admin powers when deleting DNS records\" into stable/stein\n0d514d3c15 Merge \"Don\u0027t add arp responder for non tunnel network port\" into stable/stein\n514a18ee92 Don\u0027t add arp responder for non tunnel network port\n72085466b5 Optionally use admin powers when deleting DNS records\n```\n\n@Bernard: Embdded Changes are differents from https://review.opendev.org/#/c/747219","accounts_in_message":[],"_revision_number":3},{"id":"dd3529ddc84ae95d24316ce786faa1587ce17f4b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 09:04:07.000000000","message":"Patch Set 3:\n\nAll the changes embed here looks like to fixes.","accounts_in_message":[],"_revision_number":3},{"id":"bc0bc45cd3f853639501b38ef6b013f140daaf46","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:50:59.000000000","message":"Patch Set 3: Code-Review-1\n\nNeed to update the sha to pick up the change mentioned? -1 for visibility.","accounts_in_message":[],"_revision_number":3},{"id":"6fa5ca3a4e4ecae53759a5a61b0bd5015af7f16e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-27 07:39:15.000000000","message":"Patch Set 3:\n\nYes sorry for the noise, https://review.opendev.org/#/c/740216/ is already in 14.3.0 (Jul 17 release)","accounts_in_message":[],"_revision_number":3},{"id":"779b563049b2b6116aeb053a4f257e4ef42d8ec5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-27 07:58:05.000000000","message":"Patch Set 3:\n\nIf 740216 was already in, I would recommend using 14.3.1 instead of 14.4.0.","accounts_in_message":[],"_revision_number":3},{"id":"93f0ef4ffe9f0ebec4eb845b77d48abada2ebfa9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-27 08:13:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6f594a61e3c4c6113b1582763a6e76f4e57dbc48","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-27 08:14:09.000000000","message":"Patch Set 4:\n\n\u003e Yes sorry for the noise, https://review.opendev.org/#/c/740216/ is\n \u003e already in 14.3.0 (Jul 17 release)\n\nAnd sorry that I didn\u0027t check it second time when Bernard asked me about it :)","accounts_in_message":[],"_revision_number":4},{"id":"39bf2ea67ab386b6c58983830f2f20405d998ebd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:16:11.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"05343e6b52846fac275801423b339959f887358b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-27 08:17:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"130fb918bfae372fd7844d379b2c02555c8c1ecc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 08:21:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"95899cad637bde7315f355f773bea693abe2ccd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:33:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af1f252ae5f4a71b553b8ef5e5c20d8 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4b005aaf6f9444e9ddf210406c4156b : SUCCESS in 18m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b640c74538944132af7120c487b1a2b9 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"b0284242045cb2873426e691c6ae457e5fd1d765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:33:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c41dbbecc7326848019e4c6a2c906c090b6ffdf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:05:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab46927035fc40fbbfb4db69e5719bbc : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0c1f5fcb574fc6b74521aedf086c1a : SUCCESS in 9m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/830118d14ae74a4fa0aff6d941dfc388 : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":4},{"id":"8ce8dd803231db5eae0cbfc3b2a9ad17467aecf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:05:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e713e85e040d3a9b7621ce0e9d226be71053da36","revisions":{"5716e733a504d56a262f0422387e8525fb1a76fa":{"kind":"REWORK","_number":1,"created":"2020-08-25 09:12:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/747873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/747873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/747873/1"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 09:11:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 09:11:33.000000000","tz":120},"subject":"Release neutron 14.3.1 for stable/stein","message":"Release neutron 14.3.1 for stable/stein\n\nChange-Id: Iffb08675b6911336acc5e2a7dee3665282976441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5716e733a504d56a262f0422387e8525fb1a76fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5716e733a504d56a262f0422387e8525fb1a76fa"}]},"branch":"refs/heads/master"},"24e6c9798fec82ade4095855bcbf32a5c1fe29b9":{"kind":"REWORK","_number":2,"created":"2020-08-26 08:48:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/747873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/747873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/747873/2"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 09:11:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-26 08:48:15.000000000","tz":0},"subject":"Release neutron 14.3.1 for stable/stein","message":"Release neutron 14.3.1 for stable/stein\n\nChange-Id: Iffb08675b6911336acc5e2a7dee3665282976441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/24e6c9798fec82ade4095855bcbf32a5c1fe29b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/24e6c9798fec82ade4095855bcbf32a5c1fe29b9"}]},"branch":"refs/heads/master"},"de04ce56b4b4f4517ba052cc4ff0efa7106ff5ce":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-26 08:50:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/747873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/747873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/747873/3"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 09:11:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-26 08:50:31.000000000","tz":0},"subject":"Release neutron 14.4.0 for stable/stein","message":"Release neutron 14.4.0 for stable/stein\n\nChange-Id: Iffb08675b6911336acc5e2a7dee3665282976441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de04ce56b4b4f4517ba052cc4ff0efa7106ff5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de04ce56b4b4f4517ba052cc4ff0efa7106ff5ce"}]},"branch":"refs/heads/master"},"e713e85e040d3a9b7621ce0e9d226be71053da36":{"kind":"REWORK","_number":4,"created":"2020-08-27 08:13:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/747873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/747873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/747873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/747873/4"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 09:11:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-27 08:13:34.000000000","tz":0},"subject":"Release neutron 14.3.1 for stable/stein","message":"Release neutron 14.3.1 for stable/stein\n\nChange-Id: Iffb08675b6911336acc5e2a7dee3665282976441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e713e85e040d3a9b7621ce0e9d226be71053da36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e713e85e040d3a9b7621ce0e9d226be71053da36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
