)]}'
{"id":"openstack%2Freleases~748156","triplet_id":"openstack%2Freleases~master~I38938573f105895a33aaf0a8b78cc18480539f0c","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I38938573f105895a33aaf0a8b78cc18480539f0c","subject":"nova: Release Ussuri 21.1.0","status":"MERGED","created":"2020-08-26 10:33:39.000000000","updated":"2020-08-27 08:05:31.000000000","submitted":"2020-08-27 08:05:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748156-1598515531476-d5824c7c","meta_rev_id":"3a243fbf6c8a344f6b97275e9a638c09743d051c","_number":748156,"virtual_id_number":748156,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-08-27 07:44:49.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-08-26 13:50:50.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 08:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-27 07:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-08-27 07:48:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 12:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-27 07:56:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-08-26 12:32:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 12:31:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-08-26 13:50:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-26 20:44:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 07:48:54.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-08-27 07:56:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-27 08:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cecaa101d6b5e327dd395a34be4d88521f98aba6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-26 10:33:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad8af5788e0b81cd031c39375b814cf5c0fd936c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-26 10:39:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fa1faf65dcabf81dd8bdc2c94cb543f49786d475","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:40:49.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"f9bfed9cb12d89ecbb8efbdf75b377344bb8a8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03609b5e63f64b57b05dce22428b7abe : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65763a0778b841f19fd91b7e061b6210 : SUCCESS in 6m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/32a6aa2192ae46ecbca3a90d22c1eef6 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"49e2cf082d98d8510d4cd1ec94552a9deefb7745","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-26 11:10:49.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM thanks Elod!","accounts_in_message":[],"_revision_number":1},{"id":"27f044a50bee1a82ba32f24871e3ee98c6f8fac7","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-26 11:12:23.000000000","message":"Patch Set 1: -Code-Review\n\nOps sorry did we want to include  I32908b77c18f8ec08211dd67be49bbf903611c34 that\u0027s currently in the stable/ussuri gate?","accounts_in_message":[],"_revision_number":1},{"id":"60eedcccdc54ab271da22e860e1092467c9ebafa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-26 11:15:05.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s wait until this gets merged: https://review.opendev.org/#/c/747357/","accounts_in_message":[],"_revision_number":1},{"id":"aae9986652c1017726b6ec662509914f284e7a58","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-26 11:26:36.000000000","message":"Patch Set 1: Code-Review-1\n\nwaiting for https://review.opendev.org/#/c/747357/","accounts_in_message":[],"_revision_number":1},{"id":"b76a65b6bc7926a16c556a6f919cceb05233101e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 11:27:52.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"a7def15c06002ea861364d00aa702f91c9a3f8c9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-26 12:27:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8546fc960c980e4bb6838c737964e12c6fa2ef1e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-26 12:31:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d00a8589bcc95ed1766da9cd6e67906cf57913f3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:32:56.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"d752bbcaa999ec79e2116e79292b36c20b574aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:59:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af470e6fc8640d48a3245b37731295e : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f1e17d8ebf8479581e8c6cb4df9e63f : SUCCESS in 5m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8f6b342578034aeeb7e3986c0b191364 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0e883ea623b2eabb935d50e834130d8bb80d8bb0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 13:50:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c48fdfeb3bebd07664ccc835f4fdb15d658c56b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:44:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ae58a0b8f552a1f31dc832277a7baf20b9e1e7a5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-27 07:44:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec77ab93dfdf94b3ac69ffafd60685405fee29ea","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-27 07:48:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"15bc7d17773edcdc1c4ad12387052518de7df751","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-27 07:56:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"455dbdea363677cc7ab3b678a3852f6c48df04b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 07:56:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7dd651ecffa915aed814baf5cd746a2135ca661c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:05:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a7f0b5073e449e807a134eb7e38db4 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dd9d3c80194f5a99d5084bb7a1ecb6 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f25eb0c23ef4e3bbe5ee488a061350a : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3a243fbf6c8a344f6b97275e9a638c09743d051c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:05:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2461efaad7aa755acfaa7a726f4f543ffeb70b33","revisions":{"bb01ab559e7108acdb662c68c2f009443342e042":{"kind":"REWORK","_number":1,"created":"2020-08-26 10:33:39.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/748156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/748156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/748156/1"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-08-26 10:24:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-08-26 10:33:05.000000000","tz":120},"subject":"nova: Release Ussuri 21.1.0","message":"nova: Release Ussuri 21.1.0\n\nMinor version bump required with this release due to patch\n\u0027libvirt: Mark e1000e VIF as supported\u0027.\n\n$ git log --oneline --no-merges 21.0.0..HEAD\nbbf9d1de06 libvirt: Provide VIR_MIGRATE_PARAM_PERSIST_XML during live migration\n30ee639b62 [Trivial] Remove wrong format_message() conversion\n034b2fa8fe compute: Don\u0027t delete the original attachment during pre LM rollback\n6db72002a6 Add regression tests for bug #1889108\n6070c55793 func: Add live migration rollback volume attachment tests\n726ca4aec5 func: Add _live_migrate helper to InstanceHelperMixin\n9069b2802f tests: Define constants in \u0027_IntegratedTestBase\u0027\nac05bc3b38 Handle multiple \u0027vcpusched\u0027 elements during live migrate\n42bfd50626 compute: Do not allow rescue attempts using volume snapshot images\naed86ee5d6 objects: Update keypairs when saving an instance\n93058ae1b8 libvirt: Handle VIR_ERR_DEVICE_MISSING when detaching devices\ncf399a363c zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\necd218b1a4 Correct reported system memory\n247e33af8f catch libvirt exception when nodedev not found.\n840de3b892 libvirt: Mark e1000e VIF as supported\n02f213b831 Fix cherry-pick check for merge patch\n9d3ebd58aa libvirt: Don\u0027t allow \"reserving\" file-backed memory\na6a48e876c Guard against missing image cache directory\needc0b5ff3 hardware: Raise useful error for invalid mempage size\n5cacfaab82 Check cherry-pick hashes in pep8 tox target\n8b48ca672d libvirt: Don\u0027t delete disks on shared storage during evacuate\n6ccd13f8ae Add functional test for bug 1550919\ned05246069 remove support of oslo.messaging 9.8.0 warning message\n57702ad484 Add admin doc information about image cache resource accounting\n968981b585 Reserve DISK_GB resource for the image cache\na270eeeb9b compute: Allow snapshots to be created from PAUSED volume backed instances\nfa2bfac862 Make quotas respect instance_list_per_project_cells\n1caddc4250 Silence amqp heartbeat warning\ne8b3927c92 Update scheduler instance info at confirm resize\n016eeec984 Reproduce bug 1869050\ne7ea1226b1 Bump hacking min version to 3.0.1\ne61d002530 Remove stale nested backport from InstancePCIRequests\n\nChange-Id: I38938573f105895a33aaf0a8b78cc18480539f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb01ab559e7108acdb662c68c2f009443342e042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb01ab559e7108acdb662c68c2f009443342e042"}]},"branch":"refs/heads/master"},"2461efaad7aa755acfaa7a726f4f543ffeb70b33":{"kind":"REWORK","_number":2,"created":"2020-08-26 12:27:12.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/748156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/748156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/748156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/748156/2"}}},"commit":{"parents":[{"commit":"240e8714ab8df0ef4cb3147eb3506815f88b2f3f","subject":"release a vitrage_tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/240e8714ab8df0ef4cb3147eb3506815f88b2f3f"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-08-26 10:24:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-08-26 12:26:18.000000000","tz":120},"subject":"nova: Release Ussuri 21.1.0","message":"nova: Release Ussuri 21.1.0\n\nMinor version bump required with this release due to patch\n\u0027libvirt: Mark e1000e VIF as supported\u0027.\n\n$ git log --oneline --no-merges 21.0.0..HEAD\nbbf9d1de06 libvirt: Provide VIR_MIGRATE_PARAM_PERSIST_XML during live migration\n3f3b889f4e libvirt: Do not reference VIR_ERR_DEVICE_MISSING when libvirt is \u003c v4.1.0\n30ee639b62 [Trivial] Remove wrong format_message() conversion\n034b2fa8fe compute: Don\u0027t delete the original attachment during pre LM rollback\n6db72002a6 Add regression tests for bug #1889108\n6070c55793 func: Add live migration rollback volume attachment tests\n726ca4aec5 func: Add _live_migrate helper to InstanceHelperMixin\n9069b2802f tests: Define constants in \u0027_IntegratedTestBase\u0027\nac05bc3b38 Handle multiple \u0027vcpusched\u0027 elements during live migrate\n42bfd50626 compute: Do not allow rescue attempts using volume snapshot images\naed86ee5d6 objects: Update keypairs when saving an instance\n93058ae1b8 libvirt: Handle VIR_ERR_DEVICE_MISSING when detaching devices\ncf399a363c zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\necd218b1a4 Correct reported system memory\n247e33af8f catch libvirt exception when nodedev not found.\n840de3b892 libvirt: Mark e1000e VIF as supported\n02f213b831 Fix cherry-pick check for merge patch\n9d3ebd58aa libvirt: Don\u0027t allow \"reserving\" file-backed memory\na6a48e876c Guard against missing image cache directory\needc0b5ff3 hardware: Raise useful error for invalid mempage size\n5cacfaab82 Check cherry-pick hashes in pep8 tox target\n8b48ca672d libvirt: Don\u0027t delete disks on shared storage during evacuate\n6ccd13f8ae Add functional test for bug 1550919\ned05246069 remove support of oslo.messaging 9.8.0 warning message\n57702ad484 Add admin doc information about image cache resource accounting\n968981b585 Reserve DISK_GB resource for the image cache\na270eeeb9b compute: Allow snapshots to be created from PAUSED volume backed instances\nfa2bfac862 Make quotas respect instance_list_per_project_cells\n1caddc4250 Silence amqp heartbeat warning\ne8b3927c92 Update scheduler instance info at confirm resize\n016eeec984 Reproduce bug 1869050\ne7ea1226b1 Bump hacking min version to 3.0.1\ne61d002530 Remove stale nested backport from InstancePCIRequests\n\nChange-Id: I38938573f105895a33aaf0a8b78cc18480539f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2461efaad7aa755acfaa7a726f4f543ffeb70b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2461efaad7aa755acfaa7a726f4f543ffeb70b33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
