)]}'
{"id":"openstack%2Freleases~749198","triplet_id":"openstack%2Freleases~master~Ifed5068d5762d241233e84bf419d586f9a2b39ed","project":"openstack/releases","branch":"master","topic":"oslo-ussuri-2020-09-01","hashtags":[],"change_id":"Ifed5068d5762d241233e84bf419d586f9a2b39ed","subject":"release oslo.messaging 12.1.4 - 2020-09-01","status":"MERGED","created":"2020-09-01 07:55:21.000000000","updated":"2020-09-03 10:13:24.000000000","submitted":"2020-09-03 10:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749198-1599128004885-42a114cc","meta_rev_id":"158d007f1008fd225dea606a002a05ca8bd9928f","_number":749198,"virtual_id_number":749198,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 10:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 14:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-02 09:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-03 10:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-03 10:00:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-01 13:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 14:36:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-02 09:07:13.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-03 10:00:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-03 10:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8360299c45cd2094921fd60b0d11aa8da7552ef4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 07:55:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ef64b9ef232423053edd032740170b67c3522f1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:00:31.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"26d82240bd064874b175aa58d07de4850fe9bb71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:09:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba0d3dfde8949f4a2683cb6f48b5a33 : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c813fb83c3bd41689d2220837f8e716e : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9ab3ad0a2b34a6f9a2b32165c000ad9 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c0271a2263dc145ac4f12c9a73d1b393d5d7c744","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 13:39:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"845a4c4abdbd33505a9859192d418ff16378537d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 13:40:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3c8b3fc3230df7cccf8eb44bac83e3f84307656c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 13:53:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec18fbcd23a4e008ea710a2e608a9e5 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8bd0f3bd28e4ea190688a2c59e7b705 : SUCCESS in 10m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2510e5dbeadd480c9d828b1b8f2be277 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"03b42262c964adbcea2f6d98e29cef40390c2d77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 14:36:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7842c45d08d214bf41a6f044daff7c5940e8127b","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-02 09:07:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fd2565bbdf813b82e7885924625c52281899c1a7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-03 10:00:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a0ed90a66e39fa2b84c45edc532a5ff238f4b14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:00:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"227b5d9531096b29be2b001686fb2b8ba4a285aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:13:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d3fb0c5c4b749996bce44e064e586f9b82c8ec9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:13:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ecdecbd902409d87f199cca054ba99 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3fb2a1a0353429bab53c45e7134b092 : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/593589977d2a4241a7fed9525470370a : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"946abe5524884149cd01b05b8dc2f6071ecf3bca","revisions":{"6efcb9f62ab2e2a555afc71e4f8ee12a798a49d2":{"kind":"REWORK","_number":1,"created":"2020-09-01 07:55:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/749198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/749198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/749198/1"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:51:45.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:54:25.000000000","tz":120},"subject":"release oslo.messaging 12.4.0 - 2020-09-01","message":"release oslo.messaging 12.4.0 - 2020-09-01\n\nI released a minor version because a new option have been added [1].\n\nAlthough this adds a new option, it\u0027s just a gate on a\nbug fix to ensure we don\u0027t have unintended side effects. That actually\nmakes this backport safer than if it just unconditionally enabled the\nfix.\n\n[1] https://review.opendev.org/#/c/747366/\n\nChange-Id: Ifed5068d5762d241233e84bf419d586f9a2b39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6efcb9f62ab2e2a555afc71e4f8ee12a798a49d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6efcb9f62ab2e2a555afc71e4f8ee12a798a49d2"}]},"branch":"refs/heads/master"},"946abe5524884149cd01b05b8dc2f6071ecf3bca":{"kind":"REWORK","_number":2,"created":"2020-09-01 13:39:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/98/749198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/749198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/749198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/749198/2"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 13:37:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 13:37:36.000000000","tz":120},"subject":"release oslo.messaging 12.1.4 - 2020-09-01","message":"release oslo.messaging 12.1.4 - 2020-09-01\n\nI released a bugfix version even if a new option have been added [1]\nbecause 12.2.x is already consumed by victoria.\n\nAlthough this adds a new option, it\u0027s just a gate on a\nbug fix to ensure we don\u0027t have unintended side effects. That actually\nmakes this backport safer than if it just unconditionally enabled the\nfix.\n\n[1] https://review.opendev.org/#/c/747366/\n\nChange-Id: Ifed5068d5762d241233e84bf419d586f9a2b39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/946abe5524884149cd01b05b8dc2f6071ecf3bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/946abe5524884149cd01b05b8dc2f6071ecf3bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
