)]}'
{"id":"openstack%2Freleases~749200","triplet_id":"openstack%2Freleases~master~I3f8d89d0773b71df14e112cb7f42894b22b6f33b","project":"openstack/releases","branch":"master","topic":"oslo-independent-2020-09-01","hashtags":[],"change_id":"I3f8d89d0773b71df14e112cb7f42894b22b6f33b","subject":"release pbr 5.5.0 - 2020-09-01","status":"MERGED","created":"2020-09-01 07:57:09.000000000","updated":"2020-09-01 19:42:24.000000000","submitted":"2020-09-01 19:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749200-1598989344772-941d6ed0","meta_rev_id":"79f61e66a022eb10cbe9d792f7485cd54234b774","_number":749200,"virtual_id_number":749200,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 19:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 19:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 19:31:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-01 19:29:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 19:29:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-01 19:31:10.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-01 19:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c91a790d1364a498023f6b469babbeaeb79ad26","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 07:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9ca7d74ac0ce64e150e3b20b9acff262fe4e5a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:00:52.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"36a3cfad11244a1227ff6b499e9dd5db6f4f5e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:12:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40aa9f53a21d41fba8835c0b40610915 : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbd4ffa98dde4abbb199e6b185ae8b16 : FAILURE in 6m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9ac8ceb7ab9c4af8957c956961d8852d : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f35b1bbc5601f7c4c3a06af20170555e58168629","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 12:09:35.000000000","message":"Patch Set 1:\n\nhmm I\u0027m not sure to see what\u0027s the validation error here...\n\nLocally the validation tests pass without error (only 1 warning appear in log).\n\nHere (on CI) I get this kind of error:\n\n```\nERROR: Failed to build sdist for openstack/pbr: [Errno 2] No such file or directory: \u0027tox\u0027: \u0027tox\u0027\n```","accounts_in_message":[],"_revision_number":1},{"id":"5519fdddfa981d14cb85e4e9b251af0a5378acbd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 12:21:25.000000000","message":"Patch Set 1:\n\n\u003e hmm I\u0027m not sure to see what\u0027s the validation error here...\n \u003e \n \u003e Locally the validation tests pass without error (only 1 warning\n \u003e appear in log).\n \u003e \n \u003e Here (on CI) I get this kind of error:\n \u003e \n \u003e ```\n \u003e ERROR: Failed to build sdist for openstack/pbr: [Errno 2] No such\n \u003e file or directory: \u0027tox\u0027: \u0027tox\u0027\n \u003e ```\n\n\nMore details:\n\n```\n020-09-01 08:08:44.875588 | ubuntu-bionic | Summary\n2020-09-01 08:08:44.875613 | ubuntu-bionic | -------\n2020-09-01 08:08:44.875637 | ubuntu-bionic |\n2020-09-01 08:08:44.875656 | ubuntu-bionic |\n2020-09-01 08:08:44.875675 | ubuntu-bionic | 2 warnings found\n2020-09-01 08:08:44.875699 | ubuntu-bionic | deliverables/_independent/pbr.yaml: validate_pypi_permissions: Could not determine the sdist name for openstack/pbr to check PyPI permissions: [Errno 2] No such file or directory: \u0027tox\u0027: \u0027tox\u0027\n2020-09-01 08:08:44.875723 | ubuntu-bionic | deliverables/_independent/pbr.yaml: validate_release_branch_membership: skipping descendant test for independent project, verify branch manually\n2020-09-01 08:08:44.875743 | ubuntu-bionic |\n2020-09-01 08:08:44.875761 | ubuntu-bionic |\n2020-09-01 08:08:44.875786 | ubuntu-bionic | 3 errors found\n2020-09-01 08:08:44.875806 | ubuntu-bionic | deliverables/_independent/pbr.yaml: validate_build_sdist: Failed to build sdist for openstack/pbr: [Errno 2] No such file or directory: \u0027tox\u0027: \u0027tox\u0027\n2020-09-01 08:08:44.875825 | ubuntu-bionic | deliverables/_independent/pbr.yaml: validate_pypi_readme: README check for openstack/pbr failed: [Errno 2] No such file or directory: \u0027tox\u0027: \u0027tox\u0027\n2020-09-01 08:08:44.875844 | ubuntu-bionic | deliverables/_independent/pbr.yaml: validate_tarball_base: Could not get the name of openstack/pbr for version 5.5.0: [Errno 2] No such file or directory: \u0027tox\u0027: \u0027tox\u0027\n```\n\nLocally I just have the following:\n\n```\nSummary\n-------\n\n\n1 warnings found\ndeliverables/_independent/pbr.yaml: validate_release_branch_membership: skipping descendant test for independent project, verify branch manually\n```","accounts_in_message":[],"_revision_number":1},{"id":"5afe43aa563ee060506ce6dc0f9c2d2a24ef0bde","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 14:38:13.000000000","message":"Patch Set 1:\n\nrecheck\n\nError would seem to indicate that tox isn\u0027t being installed somewhere that the job will pick it up. But we\u0027ve seen this job run and pass on other patches, so trying a recheck to see if it was just a transient issue.","accounts_in_message":[],"_revision_number":1},{"id":"2e30a09e394a0db42743fd128143461e3982734d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:55:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9a6ed0db8847f59850fed0e795c4c2 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78d67220b7844248aef0ce32471fafa8 : FAILURE in 8m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/689181f6c7e64a0390dd56dbeb0fae55 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"19a31d0a3dd335058f1f88f7903bb13705d46346","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 15:32:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4736ab88a85f8473ab1fb2b5bec81bc990f64339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:54:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca38117b53a4f08b286cc8ddfb4b233 : SUCCESS in 8m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/905eee99ec08469f8a57935d24e26e52 : SUCCESS in 8m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e7cd7d1a600a424d8f665d2bcc728530 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":2},{"id":"dcd0c9cab2848eec9ca475a7cd4fca59bcc23d8d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-01 15:55:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e9e0dcbbd196db3081a896d26e16078e326e35c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 16:24:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9e5951e9663142159ef66588249f24146efddedb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:31:39.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"d675f9129cd787781672deb30a9dfb8bca9c7237","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-01 18:30:29.000000000","message":"Patch Set 2:\n\nrecheck for new version of the job fix dependency","accounts_in_message":[],"_revision_number":2},{"id":"a4691661ccfaba1c35d23642f80fbf341687e1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:42:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fb7aff23384df69267e2d3d3c32816 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fbccf70e804425baca429f5e9cd77db : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6c4d6a2edd1c468aaec8ec61edfba008 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5166fc49aa7b132a92949abd5be408ea5d478318","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-01 18:54:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f74e6e4d0eccdf179628261e16168d910399fef3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 18:58:09.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a5867f7e83d7de02c4a3ef3229ffca1644c2ac6f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-01 19:11:46.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c7753f602e15632a54949fc05a1ad1231d364786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:27:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619c2f680349419a8e87ed1fef318ba5 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2324a231defa46108f2f20578a7a408f : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9650a824b4664c22a09d48b0810837c0 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":4},{"id":"7660c06e904919bd7601d719e7f176241be63539","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 19:29:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"743124197ee5caffca49600ae0ab91ff7023c0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:29:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ba5491dd53dc36e9311118179a773efee2837b89","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-01 19:31:10.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2cbd89f8dd23dc68631eb114da3d18c0b960b874","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:42:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b2829dab634eba35b481bf0dd7e50053b4b9651c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:42:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa7c484860d4d4b8a1561cc110535cd : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ed41b22401410e8c5a63915f976838 : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ab0c64c930e49ce95a4743c60bea37d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6a7b326b8c62fa2534904b955b71b89531807a3f","revisions":{"1a27a11796ee822fc29ef98d76482a52c2bf6ed9":{"kind":"REWORK","_number":1,"created":"2020-09-01 07:57:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/00/749200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/749200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/749200/1"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:56:51.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:56:51.000000000","tz":120},"subject":"release pbr 5.5.0 - 2020-09-01","message":"release pbr 5.5.0 - 2020-09-01\n\nChange-Id: I3f8d89d0773b71df14e112cb7f42894b22b6f33b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a27a11796ee822fc29ef98d76482a52c2bf6ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a27a11796ee822fc29ef98d76482a52c2bf6ed9"}]},"branch":"refs/heads/master"},"095860ce54c38ca7812b8a00bfae52dd2b84f459":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-01 15:32:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/749200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/749200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/749200/2"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:56:51.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-01 15:32:28.000000000","tz":0},"subject":"release pbr 5.5.0 - 2020-09-01","message":"release pbr 5.5.0 - 2020-09-01\n\nDepends-on: https://review.opendev.org/749345\n\nChange-Id: I3f8d89d0773b71df14e112cb7f42894b22b6f33b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/095860ce54c38ca7812b8a00bfae52dd2b84f459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/095860ce54c38ca7812b8a00bfae52dd2b84f459"}]},"branch":"refs/heads/master"},"584a8817f4eb46b50b5d74a0d2d51fa2ee0a5c2f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-01 18:58:09.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/749200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/749200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/749200/3"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:56:51.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-01 18:57:32.000000000","tz":-300},"subject":"release pbr 5.5.0 - 2020-09-01","message":"release pbr 5.5.0 - 2020-09-01\n\nDepends-on: https://review.opendev.org/749366\n\nChange-Id: I3f8d89d0773b71df14e112cb7f42894b22b6f33b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/584a8817f4eb46b50b5d74a0d2d51fa2ee0a5c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/584a8817f4eb46b50b5d74a0d2d51fa2ee0a5c2f"}]},"branch":"refs/heads/master"},"6a7b326b8c62fa2534904b955b71b89531807a3f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-01 19:11:46.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/00/749200/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/749200/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/749200/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/749200/4"}}},"commit":{"parents":[{"commit":"38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd","subject":"Merge \"Add Wallaby PTG to the release schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccaee0a8179780f8f8fccb9a7fc9874fbcb3bd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:56:51.000000000","tz":120},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-01 19:11:46.000000000","tz":0},"subject":"release pbr 5.5.0 - 2020-09-01","message":"release pbr 5.5.0 - 2020-09-01\n\nChange-Id: I3f8d89d0773b71df14e112cb7f42894b22b6f33b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a7b326b8c62fa2534904b955b71b89531807a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a7b326b8c62fa2534904b955b71b89531807a3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
