)]}'
{"id":"openstack%2Freleases~749778","triplet_id":"openstack%2Freleases~master~Ifa930c53f404ad233e578a4c37f9e55717a0535e","project":"openstack/releases","branch":"master","topic":"neutron-libs-victoria","hashtags":[],"change_id":"Ifa930c53f404ad233e578a4c37f9e55717a0535e","subject":"Release neutron non-client libraries for Victoria","status":"MERGED","created":"2020-09-03 16:20:13.000000000","updated":"2020-09-04 13:03:02.000000000","submitted":"2020-09-04 13:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749778-1599224582966-1d2acb16","meta_rev_id":"29b2f8f5e0be8eefdba76f378f56dba6d14f7da5","_number":749778,"virtual_id_number":749778,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 13:03:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 12:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-04 05:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-09-04 10:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 12:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-04 05:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 05:51:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-04 10:38:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 12:50:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-04 13:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07c940cbd21b411b318d8deca9a60fc2aa5294f3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-03 16:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761e28f9c63fb6f68bcffd7f5d604a3f1627d57f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:21:34.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cd7e6322cadf25d88ec3fc820c3c7be1eb0610dd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-03 16:21:38.000000000","message":"Patch Set 1: Workflow-1\n\nWe are waiting for the following neutron-lib changes\n- https://review.opendev.org/#/c/748279/\n- https://review.opendev.org/#/c/741452/","accounts_in_message":[],"_revision_number":1},{"id":"834bf99ff5b184ddc776554e849d3edd3abf0a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:41:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9d8d0f9433437ba00f8070abe777f4 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b378828114bc4363aedee2e17b317cda : SUCCESS in 6m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/61203e391cca4b939fc92e37d41566e3 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d50ec3303d1dc79f7f91eaa90900c3f4d685218e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 05:50:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f558911a4e4b3257790e2f12472951bb32524f55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 05:51:11.000000000","message":"Patch Set 2: Code-Review+1\n\nNow all patches are merged so we should be good to go","accounts_in_message":[],"_revision_number":2},{"id":"7b47557272c4c3a0f0ddb1cf826effd46f3d2cc8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 05:51:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"33df1a7b08788179cc4f5e86e2db39be222221eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 06:01:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1165f597fca42e791850617be2d1207 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c9d62524be94f3088e0ad9f7b931fbf : SUCCESS in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64ea469c8e8942ed8a2541f30e5cb289 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1908b56364764b86bc8e609c3a02ab9feda2f885","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 10:38:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLGTM with some minor thoughts for os-ken, please see my inline comment.","accounts_in_message":[],"_revision_number":2},{"id":"2fb4a310fa5ee13a4e7a2bccbf500d9e1132b9f1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 12:50:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nWould be good to have the stable/victoria branch creation included, but we can do that in a follow up.","accounts_in_message":[],"_revision_number":2},{"id":"ceb65a20ca25bcd30d766411412cc97eb2c2a315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:50:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"245b2030136e46f9b76060be0e262f8ac919daee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:03:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c2aa1995574834943c229cb3ab796c : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0ef8f364f44583ad97355f398c1c10 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1723372e88a423aa421dc7926cf4a0f : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"99f412d59c4a8c991e2dc3f11699afed1427e1be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:03:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5c39884c14025c553cc19132766a546ab317969","revisions":{"ae94de46237ee2286ce3854dfc9f7c593be6fa13":{"kind":"REWORK","_number":1,"created":"2020-09-03 16:20:13.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/78/749778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/749778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/749778/1"}}},"commit":{"parents":[{"commit":"d6e008dd611778797d72c003946cc166ddde9a20","subject":"Merge \"Victoria Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6e008dd611778797d72c003946cc166ddde9a20"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-03 16:19:00.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-03 16:19:00.000000000","tz":540},"subject":"Release neutron non-client libraries for Victoria","message":"Release neutron non-client libraries for Victoria\n\nChange-Id: Ifa930c53f404ad233e578a4c37f9e55717a0535e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae94de46237ee2286ce3854dfc9f7c593be6fa13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae94de46237ee2286ce3854dfc9f7c593be6fa13"}]},"branch":"refs/heads/master"},"a5c39884c14025c553cc19132766a546ab317969":{"kind":"REWORK","_number":2,"created":"2020-09-04 05:50:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/749778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/749778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/749778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/749778/2"}}},"commit":{"parents":[{"commit":"d6e008dd611778797d72c003946cc166ddde9a20","subject":"Merge \"Victoria Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6e008dd611778797d72c003946cc166ddde9a20"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-03 16:19:00.000000000","tz":540},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-04 05:46:19.000000000","tz":0},"subject":"Release neutron non-client libraries for Victoria","message":"Release neutron non-client libraries for Victoria\n\nChange-Id: Ifa930c53f404ad233e578a4c37f9e55717a0535e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5c39884c14025c553cc19132766a546ab317969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5c39884c14025c553cc19132766a546ab317969"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
