)]}'
{"id":"openstack%2Freleases~749786","triplet_id":"openstack%2Freleases~master~Iefb6a401b9df100d301a0fb7bb94641b8b07fe19","project":"openstack/releases","branch":"master","topic":"victoria-lib-freeze","hashtags":[],"change_id":"Iefb6a401b9df100d301a0fb7bb94641b8b07fe19","subject":"Release final metalsmith for Victoria","status":"MERGED","created":"2020-09-03 16:49:05.000000000","updated":"2020-09-04 12:59:46.000000000","submitted":"2020-09-04 12:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749786-1599224386655-b29784c5","meta_rev_id":"5dbdc04fdd8c459f735eb624f5d3198205b43af7","_number":749786,"virtual_id_number":749786,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 12:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 12:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-04 10:33:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-09-04 10:57:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 10:47:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 12:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-04 10:49:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 16:56:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-09-03 16:56:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-04 10:33:16.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-04 10:47:30.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-09-04 10:57:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 12:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3135deb4d6407cba580d00a3135738f0c023d43f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-03 16:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd163ab8455c1abc72c32382be3234a453e98c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56dac3d02efd4b75ad5d65281e1281e6 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd44467b93ad47d0be058de6022e9dfd : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7926a50c0f444fc9c5de684fe03f6a6 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0746aa1837ac3f03067386641313f0fd4427ef25","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-04 08:49:03.000000000","message":"Patch Set 1: Code-Review-1\n\nWe\u0027re good to go, but that\u0027s a feature release, so should be 1.2.0. Thank you!","accounts_in_message":[],"_revision_number":1},{"id":"50cb6ccbd4b07221f2c2a8d484dbce3bb06848c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 09:00:30.000000000","message":"Patch Set 1:\n\n\u003e We\u0027re good to go, but that\u0027s a feature release, so should be 1.2.0.\n \u003e Thank you!\n\nWhich feature you refer to?\nAt first glance a bugfix version LGTM but I surely missed something.","accounts_in_message":[],"_revision_number":1},{"id":"56ff80fd50f3269b63a9a7f90427d973a1f9739c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-04 09:06:05.000000000","message":"Patch Set 1:\n\nhttps://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c and follow-up commits for it","accounts_in_message":[],"_revision_number":1},{"id":"0c59fcb60aea54dc7768e1062a7c653f74dee583","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 09:14:09.000000000","message":"Patch Set 1:\n\n\u003e https://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c\n \u003e and follow-up commits for it\n\nAck thanks, I was wondering if it was this one, then I agree with you that a minor version should be released.","accounts_in_message":[],"_revision_number":1},{"id":"f0b253140c4db919e30c4fcbf61aa9619f45f5eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 09:15:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f50ed72ffb57d207b5e565a1872b071f147068f","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 09:21:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0318ca95917c9b767d05590b85234933d47f68bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:27:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382993cc75cd4814a027094fabc8b0c6 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a41c7eda47b04c0eb3389a89c768fd6a : FAILURE in 5m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8e7f5a1342ca4fb096ac7d450ea075d7 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8fc5760926a5c3869024932e95c8e15bae8d7bd0","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 10:12:11.000000000","message":"Patch Set 2:\n\nJust for me to understand: The issue in CI is a mismatch between initial proposal to tag 1.1.1 whereas new one is 1.2.0 (since feat).\nIs it a \"regular\" issue in this specific case or something needs to be done ?","accounts_in_message":[],"_revision_number":2},{"id":"1ebcba3dd35f7163cea8d60bd53a4d29710ef729","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-04 10:16:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nSee inline","accounts_in_message":[],"_revision_number":2},{"id":"f5b706d7926f9d91a7250d7e8938bf00d74b4c1e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 10:20:28.000000000","message":"Removed Code-Review+1 by Boyron Sébastien \u003csebastien@boyron.eu\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"84ae1411f526fe8bc22417d3e17022ba21bb68fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 10:30:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9540aedcf7e49eb5d3f74d7e0ec8bfdea19254c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 10:31:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e7b14b2c9dcc72c49020ca298aad82bb6525032e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 10:33:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bd965eb4f2da383a272609a780fb313530e66635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:44:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fcf61f9c0a4e2988afacbaf4314a10 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b90e276e2c49425dbdd47570d1050b66 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d89412d9cdf442af9ee42789a458e600 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2dd8c8a5370b3ea3f391ce58b07e2fec1450f748","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-04 10:47:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"722405dc4ca9d024d3f89b948219407498819635","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:49:05.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"07a51698c51e5102fd819c7fc8aded041529e354","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 10:57:06.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM for the current version","accounts_in_message":[],"_revision_number":3},{"id":"b0fb5cbfa1bf0439b82b246ce65a20f4e70e0d01","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 12:48:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5bd38681c2d975554357bcbdf3a66068be4dad70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:49:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"858f2b06e691f452b8e6318edd9c893138d2ec8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:59:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fca756dce2548389d09538ed61ec1f2 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae24061811454eccab299be392d0048e : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e55924ef81ce4a1a8b73d3f99afce312 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0fb823514bde97e766c51419339cd061409dce7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:59:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3869ef2c8fb0bf17e24d80a5ed73076219b2e4df","revisions":{"90160a84c5ddfb99e8af30a2821a08161c40488c":{"kind":"REWORK","_number":1,"created":"2020-09-03 16:49:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/86/749786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/749786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/749786/1"}}},"commit":{"parents":[{"commit":"005e926d80bfe747079cfdd369199099f644692e","subject":"Release ironic-lib and create stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/005e926d80bfe747079cfdd369199099f644692e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 16:49:04.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 16:49:04.000000000","tz":-300},"subject":"Release final metalsmith for Victoria","message":"Release final metalsmith for Victoria\n\nThis proposed a release for metalsmith to get any final merged changes\nout prior to the lib freeze for Victoria going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, Sept 7.\n\nChange-Id: Iefb6a401b9df100d301a0fb7bb94641b8b07fe19\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90160a84c5ddfb99e8af30a2821a08161c40488c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90160a84c5ddfb99e8af30a2821a08161c40488c"}]},"branch":"refs/heads/master"},"10c93d3ffd936b7122e9cb7510483bf8fab5608c":{"kind":"REWORK","_number":2,"created":"2020-09-04 09:15:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/749786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/749786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/749786/2"}}},"commit":{"parents":[{"commit":"005e926d80bfe747079cfdd369199099f644692e","subject":"Release ironic-lib and create stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/005e926d80bfe747079cfdd369199099f644692e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 16:49:04.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-04 09:15:35.000000000","tz":120},"subject":"Release final metalsmith for Victoria","message":"Release final metalsmith for Victoria\n\nThis proposed a release for metalsmith to get any final merged changes\nout prior to the lib freeze for Victoria going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, Sept 7.\n\nChange-Id: Iefb6a401b9df100d301a0fb7bb94641b8b07fe19\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10c93d3ffd936b7122e9cb7510483bf8fab5608c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10c93d3ffd936b7122e9cb7510483bf8fab5608c"}]},"branch":"refs/heads/master"},"3869ef2c8fb0bf17e24d80a5ed73076219b2e4df":{"kind":"REWORK","_number":3,"created":"2020-09-04 10:31:05.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/86/749786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/749786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/749786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/749786/3"}}},"commit":{"parents":[{"commit":"005e926d80bfe747079cfdd369199099f644692e","subject":"Release ironic-lib and create stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/005e926d80bfe747079cfdd369199099f644692e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 16:49:04.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-04 10:30:55.000000000","tz":120},"subject":"Release final metalsmith for Victoria","message":"Release final metalsmith for Victoria\n\nThis proposed a release for metalsmith to get any final merged changes\nout prior to the lib freeze for Victoria going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, Sept 7.\n\nChange-Id: Iefb6a401b9df100d301a0fb7bb94641b8b07fe19\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3869ef2c8fb0bf17e24d80a5ed73076219b2e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3869ef2c8fb0bf17e24d80a5ed73076219b2e4df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
