)]}'
{"id":"openstack%2Freleases~750428","triplet_id":"openstack%2Freleases~master~Idac43984cdd093f4b6933871a58b36959cea1f94","project":"openstack/releases","branch":"master","topic":"victoria-client-lib","hashtags":[],"change_id":"Idac43984cdd093f4b6933871a58b36959cea1f94","subject":"Release final python-neutronclient for victoria","status":"MERGED","created":"2020-09-08 15:28:22.000000000","updated":"2020-09-09 16:29:39.000000000","submitted":"2020-09-09 16:29:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750428-1599668979472-9fd2f9ed","meta_rev_id":"97703695ef98609a5ff6552bd3e19a158a0c315b","_number":750428,"virtual_id_number":750428,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 16:29:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-09 16:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-09 07:53:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-09-09 13:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-09-09 07:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-09 16:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-09 07:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 07:19:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-09 07:53:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-09 13:44:35.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-09 16:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"132da0a8c055301d4782b5872b724644eebf0ba7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-08 15:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ea08b6db68cdb3d524c6ca3539de3b7e9490b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 16:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497f2f89f4e645c480d527b20dd21f41 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a6fb52e3d6e4d2c9f536d47bad95e5f : SUCCESS in 8m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2a526049f2bf4d8d8dffa51e23a9af48 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ffb6e7cd40051cfbac253b9a04682d0e85a4bc27","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 16:24:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fbaf90f80e92cd3d3ff6e936bc87850ad057034","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-09 07:10:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"34a47322e527a9764079867a563aaf978d510cf7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-09 07:11:19.000000000","message":"Patch Set 2:\n\nI updated hash to the latest one and now it should be good to go. But my question is: shouldn\u0027t we bump minor version here? Just in case if we would need to e.g. release some bug fix version of stable/ussuri at some point.","accounts_in_message":[],"_revision_number":2},{"id":"515139ee8b7bf82296d99232ffd915f2084fbe23","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 07:15:23.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"68fada823c9f2014020442951f3ec68fa836c626","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 07:19:54.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"3f0d6b874e39d8727fad5d06495e305924786fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 07:24:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72c0684aff84b41837fd8af99288b69 : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad7bda127ffb424fbba464857eda1b8b : SUCCESS in 8m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1b1f2cc686a741ca86822987707f60a5 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d106ddd010e613655b9aefffd9d272f65b7e3f6b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 07:25:35.000000000","message":"Patch Set 2:\n\n\u003e I updated hash to the latest one and now it should be good to go.\n \u003e But my question is: shouldn\u0027t we bump minor version here? Just in\n \u003e case if we would need to e.g. release some bug fix version of\n \u003e stable/ussuri at some point.\n\nI don\u0027t seen reason to release a minor version with these changes,  ussuri is already stuck on 7.1.x as 7.2.x is already consumed by victoria and already have been released, though.\n\nPlease let me know if I answer your question.","accounts_in_message":[],"_revision_number":2},{"id":"2e3d1c89ecac4535c215f1a78d1b8f6a3e1275b9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-09 07:53:38.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e \u003e I updated hash to the latest one and now it should be good to go.\n \u003e \u003e But my question is: shouldn\u0027t we bump minor version here? Just in\n \u003e \u003e case if we would need to e.g. release some bug fix version of\n \u003e \u003e stable/ussuri at some point.\n \u003e \n \u003e I don\u0027t seen reason to release a minor version with these changes, \n \u003e ussuri is already stuck on 7.1.x as 7.2.x is already consumed by\n \u003e victoria and already have been released, though.\n \u003e \n \u003e Please let me know if I answer your question.\n\nSorry, I didn\u0027t check that ussuri is using 7.1.x, not 7.2 :) So all good for me :)","accounts_in_message":[],"_revision_number":2},{"id":"7c97fffe9cc813e3bbba29c91864cc5f9f0d3df7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-09 13:44:35.000000000","message":"Patch Set 2: Code-Review+1\n\nYeah, it is a good practice to increment the minor version when we release a first deliverable for a new release (even if we don\u0027t have changes worth for a new minor version) so that a previous stable branch can release a new one in that branch. We already do this in the first release of python-neutronclient (7.2.0).","accounts_in_message":[],"_revision_number":2},{"id":"d5122db308f70da3632f25817d5d506a35633619","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-09 16:05:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23e5cd5ec02075bbb506c2151f2ca13320753efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 16:14:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd9dee9334f2d7156b9e8c36cb5e35ea89d5fcc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 16:29:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ae44fd0aa81f4112c00b51f03f3453c7f395a187","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 16:29:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8537e6afffa943c6909fca2aeb93d961 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e0101b411946f8a572109acfd3aaf0 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da24cff74ec047c2b45ffb9252e588f6 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06edffc4ba53c7a04ee8aae45685f838f10b30f2","revisions":{"e40dfbe2af65e05e501609a222e31463bf13d3f2":{"kind":"REWORK","_number":1,"created":"2020-09-08 15:28:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/28/750428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/750428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/750428/1"}}},"commit":{"parents":[{"commit":"ac3f800241eca9588e555019b987956a43531f06","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac3f800241eca9588e555019b987956a43531f06"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:28:20.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:28:20.000000000","tz":-300},"subject":"Release final python-neutronclient for victoria","message":"Release final python-neutronclient for victoria\n\nFinal victoria client lib release for the September 10 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave a -1 if we\nneed to wait a little bit longer for something final to loand,\nthen update with a better commit hash to use for the release when\nready.\n\nPatches with no response will be approved by September 14.\n\nChange-Id: Idac43984cdd093f4b6933871a58b36959cea1f94\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e40dfbe2af65e05e501609a222e31463bf13d3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e40dfbe2af65e05e501609a222e31463bf13d3f2"}]},"branch":"refs/heads/master"},"06edffc4ba53c7a04ee8aae45685f838f10b30f2":{"kind":"REWORK","_number":2,"created":"2020-09-09 07:10:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/28/750428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/750428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/750428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/750428/2"}}},"commit":{"parents":[{"commit":"ac3f800241eca9588e555019b987956a43531f06","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac3f800241eca9588e555019b987956a43531f06"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:28:20.000000000","tz":-300},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-09 07:09:47.000000000","tz":0},"subject":"Release final python-neutronclient for victoria","message":"Release final python-neutronclient for victoria\n\nFinal victoria client lib release for the September 10 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave a -1 if we\nneed to wait a little bit longer for something final to loand,\nthen update with a better commit hash to use for the release when\nready.\n\nPatches with no response will be approved by September 14.\n\nChange-Id: Idac43984cdd093f4b6933871a58b36959cea1f94\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06edffc4ba53c7a04ee8aae45685f838f10b30f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06edffc4ba53c7a04ee8aae45685f838f10b30f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
