)]}'
{"id":"openstack%2Freleases~750436","triplet_id":"openstack%2Freleases~master~I98c93bf0806990d3cc3ada045606627fda222891","project":"openstack/releases","branch":"master","topic":"victoria-client-lib","hashtags":[],"change_id":"I98c93bf0806990d3cc3ada045606627fda222891","subject":"Release final python-tackerclient for victoria","status":"MERGED","created":"2020-09-08 15:30:43.000000000","updated":"2020-09-16 21:42:33.000000000","submitted":"2020-09-16 21:42:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750436-1600292553602-b2ad87a5","meta_rev_id":"d24f407b485eeff05b81d804476b063aa981f5fc","_number":750436,"virtual_id_number":750436,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 21:42:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-16 21:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2020-09-16 19:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-16 21:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 00:48:44.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-09-16 19:34:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-16 21:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbcd231784bc2983ff01eb17f3cf84f2aad45ad1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-08 15:30:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b8440e1e2bcf790ad0c8b48715c7d11b686fe81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 16:43:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501855539bc14ca9b2d093593b0719fc : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa7f0fc74a644529a2a9e54c13cbf018 : SUCCESS in 7m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ffa430fdfe5e4d9ba182347ee0668551 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7ffa4a5df7a782fbd4af849bcab46c4709628524","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 16:55:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92f15d81de33b94c028fc22b50ff258f305d8c34","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-09 06:39:25.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease wait for some patches still under reviewing.","accounts_in_message":[],"_revision_number":1},{"id":"18564c181913685d33cbbbee50b041a21a326510","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 06:50:06.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"aa46623ed5736ba63da01d88ae92cfaaeccb36f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-14 13:21:55.000000000","message":"Patch Set 1:\n\nIs this ready Yasufumi? We are past the deadline for client library freeze.","accounts_in_message":[],"_revision_number":1},{"id":"49590319e7be6610f9a203856e273be3558e32a1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 13:38:50.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM in the current state","accounts_in_message":[],"_revision_number":1},{"id":"ea25a9ee5430dbd8177653a866e70db872626800","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-14 14:46:14.000000000","message":"Patch Set 1:\n\nHi Hervé,\nWe have still three patches waiting for another +2 actually. I\u0027ll ask cores to give it quickly.","accounts_in_message":[],"_revision_number":1},{"id":"48892fec7f349625ff24a325a8302ae246bb42dd","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-15 10:24:37.000000000","message":"Patch Set 1:\n\nHi,\n\nSorry for waiting for a long time. We\u0027ve completed to review all patches and run workflow, but cannot be merged because of some trouble in zuul itself. So, could you please wait for a while to be fixed the serious problem?","accounts_in_message":[],"_revision_number":1},{"id":"089c4bb5affd0a1f90f8f9e468c49cb23de3ac99","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-15 12:59:30.000000000","message":"Patch Set 1:\n\nWe are already 5 days past the client lib freeze. It might be good to consider whether something is too late for this release cycle and push it out to the wallaby release.\n\nWe can wait a couple more days though.","accounts_in_message":[],"_revision_number":1},{"id":"0027703e5de4a88a5fff0f16470c55cd98bb8bc1","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-16 00:48:44.000000000","message":"Patch Set 1: Code-Review+1\n\nHi Sean, I apologize for waiting so long. Zuul was fixed and we could merge the patches.","accounts_in_message":[],"_revision_number":1},{"id":"eed29bb43c2c3f68a7f44d6210d0267d7673a48e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:49:46.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b00d5a8efcdd84ad4738b0cdd7235f41ccb0996e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 14:08:16.000000000","message":"Patch Set 1:\n\n\u003e Hi Sean, I apologize for waiting so long. Zuul was fixed and we\n \u003e could merge the patches.\n\nDid you want to update the commit sha to pick up any new merges?","accounts_in_message":[],"_revision_number":1},{"id":"5d21c2b4f37185421440259e0c00d5293598a026","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-16 18:43:29.000000000","message":"Patch Set 1:\n\n\u003e Did you want to update the commit sha to pick up any new merges?\n\nYes, if it\u0027s possible to pick up the latest five updates merged after 11th Sept. It\u0027s the first time of my PTL role, so please let me know if I ask you such a something wrong. Thank.\n\n- https://review.opendev.org/#/c/747374/\n- https://review.opendev.org/#/c/751964/\n- https://review.opendev.org/#/c/750583/\n- https://review.opendev.org/#/c/740060/\n- https://review.opendev.org/#/c/748971/","accounts_in_message":[],"_revision_number":1},{"id":"efdf2265f0e0846d55c6e19ef3ea70eeb8ecaadd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 18:46:44.000000000","message":"Patch Set 1:\n\nYes, feel free to edit this patch to change it to the latest commit sha from the repo.\n\nI can try to get to it a little later if you are not able to.","accounts_in_message":[],"_revision_number":1},{"id":"da8a2589b48894da87bac975b6e9ccd7de341395","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 19:12:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f4b353e9bf2763c9a76b7865de9f0cb907155048","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 19:21:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nWaiting for project team\u0027s feedback","accounts_in_message":[],"_revision_number":2},{"id":"8a1bdd479474566ab70da4f3aed442ec301c2501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:27:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4368f67597e54bd48472365eb55c48ef : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d00e170dbe94c66bd9f15fb63df5f8a : SUCCESS in 7m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cb54f8a07f4c4a5ab11dc0adec21c781 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e1b1a70a1f32882fbaf6dc4104ad1e2e5324ad06","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 19:30:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3637486b937ce4132d4a34f4bd479281f51dc0d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 19:31:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b3182356cda2eda3f00beed5794804c95d88a77","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 19:34:14.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"768912bff76063b641cde55b439ef67e2ff27220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:44:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af8fa9016c24c34bf9790e86fbb806a : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbe96a486c0c4c7da66c494231e63ea3 : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5e7feb72c26845a0b31f9d3d3142fcde : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7881f409ceac334e0af0d77cd96b9f9cf718633a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 21:31:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d128bf4fd062406007debf1be23383bb254a59c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:31:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f6fb5d9ff5e7c85d9c3ba9f4ad04074d3186b4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:42:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d66a6e32a9fa69f1cb72d26e690a3a31a13de979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:42:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1785ffa924a848efabe3fcd82df6ba09 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f861ff8a58e94f439b97478cf6c95186 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f6976add6df4af39e31cf1d7cda2bbd : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c5e45da900aced1b69975d34aed181096df144a4","revisions":{"a37f2442ebf631ce0b9477919d9313647e358d03":{"kind":"REWORK","_number":1,"created":"2020-09-08 15:30:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/750436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/750436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/750436/1"}}},"commit":{"parents":[{"commit":"ac3f800241eca9588e555019b987956a43531f06","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac3f800241eca9588e555019b987956a43531f06"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:30:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:30:41.000000000","tz":-300},"subject":"Release final python-tackerclient for victoria","message":"Release final python-tackerclient for victoria\n\nFinal victoria client lib release for the September 10 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave a -1 if we\nneed to wait a little bit longer for something final to loand,\nthen update with a better commit hash to use for the release when\nready.\n\nPatches with no response will be approved by September 14.\n\nChange-Id: I98c93bf0806990d3cc3ada045606627fda222891\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37f2442ebf631ce0b9477919d9313647e358d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37f2442ebf631ce0b9477919d9313647e358d03"}]},"branch":"refs/heads/master"},"98ad9b46a3b2ac8c454aaa6eb0da61005f957ab0":{"kind":"REWORK","_number":2,"created":"2020-09-16 19:12:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/750436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/750436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/750436/2"}}},"commit":{"parents":[{"commit":"ac3f800241eca9588e555019b987956a43531f06","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac3f800241eca9588e555019b987956a43531f06"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:30:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-16 19:12:03.000000000","tz":0},"subject":"Release final python-tackerclient for victoria","message":"Release final python-tackerclient for victoria\n\nFinal victoria client lib release for the September 10 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave a -1 if we\nneed to wait a little bit longer for something final to loand,\nthen update with a better commit hash to use for the release when\nready.\n\nPatches with no response will be approved by September 14.\n\nChange-Id: I98c93bf0806990d3cc3ada045606627fda222891\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98ad9b46a3b2ac8c454aaa6eb0da61005f957ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98ad9b46a3b2ac8c454aaa6eb0da61005f957ab0"}]},"branch":"refs/heads/master"},"c5e45da900aced1b69975d34aed181096df144a4":{"kind":"REWORK","_number":3,"created":"2020-09-16 19:30:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/750436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/750436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/750436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/750436/3"}}},"commit":{"parents":[{"commit":"ac3f800241eca9588e555019b987956a43531f06","subject":"Merge \"Release tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac3f800241eca9588e555019b987956a43531f06"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-08 15:30:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-16 19:30:41.000000000","tz":0},"subject":"Release final python-tackerclient for victoria","message":"Release final python-tackerclient for victoria\n\nFinal victoria client lib release for the September 10 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave a -1 if we\nneed to wait a little bit longer for something final to loand,\nthen update with a better commit hash to use for the release when\nready.\n\nPatches with no response will be approved by September 14.\n\nChange-Id: I98c93bf0806990d3cc3ada045606627fda222891\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5e45da900aced1b69975d34aed181096df144a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5e45da900aced1b69975d34aed181096df144a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
