)]}'
{"id":"openstack%2Freleases~751276","triplet_id":"openstack%2Freleases~master~If64808a25800a5bcee7075a06048ef4e1a9757f2","project":"openstack/releases","branch":"master","topic":"Doc-check-independent-libs-cmd","hashtags":[],"change_id":"If64808a25800a5bcee7075a06048ef4e1a9757f2","subject":"Improve the command to list libs that could be candidates for independent release model","status":"MERGED","created":"2020-09-11 13:29:30.000000000","updated":"2020-09-14 09:19:39.000000000","submitted":"2020-09-14 09:19:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751276-1600075179056-e73e0a8c","meta_rev_id":"442dd25e8c2dc909996a99794494ca0e5fb34f88","_number":751276,"virtual_id_number":751276,"owner":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 09:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-14 09:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-09-11 13:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-14 09:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 13:50:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 13:56:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-14 09:03:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-14 09:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7693b090d38dfaefe1fb2bf29ae1013d14b49c42","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 13:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"521c43c4f555e69b391f5e18c6e2cedc3261dc4d","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 13:35:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f23aa5a45302d765a58fa2ed08912a87fb976269","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 13:41:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7216704b75eb55b2781b7b90d75d97a24fe1517d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 13:43:54.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM in the current state, my previous comment is a suggestion","accounts_in_message":[],"_revision_number":2},{"id":"f76028f2053624f3aba13f1bccf9dc2583795873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:44:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b999650307443e199d5b33840fe4c05 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f98bdcd1b4e0e9407e586de421e807d0847dc594","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 13:50:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThanks Sebastien, I think this is a good improvement. Will wait a bit to give Thierry a chance to see it, just in case he remembers some reason we didn\u0027t use this command in the first place.","accounts_in_message":[],"_revision_number":2},{"id":"ff4971516c48ee7975b40d2d390ed3b8e017eff1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 13:51:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86e6f1198d6df838d1395ee070fab45a0a186dfa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 13:52:06.000000000","message":"Patch Set 2:\n\nElse the command seems to work as expected :)","accounts_in_message":[],"_revision_number":2},{"id":"de17058c000f183f145bb2ec09f640f1892e7a58","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 13:54:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c42d691cc05d9879e2a09dfcd34836fffca54ce","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 13:56:10.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks :)","accounts_in_message":[],"_revision_number":3},{"id":"857039832a6858adf3477e4843eb831c3558f849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 14:06:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c86a2ae13eb4bbbb772c27cbf788919 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":3},{"id":"15d7387ed27294337767e65871e7a11fa3612b73","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-14 09:03:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c15cbe6da43cd6be77ef346ffb7b4ee5a9eab41d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:03:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d48563991f332e047474c3fb4969b798653f467f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:19:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0e81f23eda4100b322dd5cfc8d5192 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c53c4004c3044809472e420f3262e21 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4dcfd1f8d091afd59a8d5cd69ad989e0b16a3ef3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:19:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2d964b4c984d85ff227d888dd93b9f21e10d3128","revisions":{"491600dabbe5ff93349c41309db6ca282d1ad66f":{"kind":"REWORK","_number":1,"created":"2020-09-11 13:29:30.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/76/751276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/751276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/751276/1"}}},"commit":{"parents":[{"commit":"1f098dc614c832014a0f2388162dc42522f188cd","subject":"Merge \"Add release highlights for ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f098dc614c832014a0f2388162dc42522f188cd"}]}],"author":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:11:13.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:21:04.000000000","tz":120},"subject":"Improve the command to list libs that could be candidates for independent release model","message":"Improve the command to list libs that could be candidates for independent release model\n\nChange-Id: If64808a25800a5bcee7075a06048ef4e1a9757f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/491600dabbe5ff93349c41309db6ca282d1ad66f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/491600dabbe5ff93349c41309db6ca282d1ad66f"}]},"branch":"refs/heads/master"},"c183c904e3602f8eec353202b541edc62bf14e41":{"kind":"NO_CHANGE","_number":2,"created":"2020-09-11 13:35:23.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/76/751276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/751276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/751276/2"}}},"commit":{"parents":[{"commit":"1f098dc614c832014a0f2388162dc42522f188cd","subject":"Merge \"Add release highlights for ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f098dc614c832014a0f2388162dc42522f188cd"}]}],"author":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:11:13.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:35:01.000000000","tz":120},"subject":"Improve the command to list libs that could be candidates for independent release model","message":"Improve the command to list libs that could be candidates for independent release model\n\nChange-Id: If64808a25800a5bcee7075a06048ef4e1a9757f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c183c904e3602f8eec353202b541edc62bf14e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c183c904e3602f8eec353202b541edc62bf14e41"}]},"branch":"refs/heads/master"},"2d964b4c984d85ff227d888dd93b9f21e10d3128":{"kind":"REWORK","_number":3,"created":"2020-09-11 13:54:41.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/76/751276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/751276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/751276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/751276/3"}}},"commit":{"parents":[{"commit":"1f098dc614c832014a0f2388162dc42522f188cd","subject":"Merge \"Add release highlights for ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f098dc614c832014a0f2388162dc42522f188cd"}]}],"author":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:11:13.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 13:54:04.000000000","tz":120},"subject":"Improve the command to list libs that could be candidates for independent release model","message":"Improve the command to list libs that could be candidates for independent release model\n\nChange-Id: If64808a25800a5bcee7075a06048ef4e1a9757f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d964b4c984d85ff227d888dd93b9f21e10d3128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d964b4c984d85ff227d888dd93b9f21e10d3128"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
